)]}'
{"id":"openstack%2Freleases~751409","triplet_id":"openstack%2Freleases~master~I6ec9fa77baab58df93bdadc0ac3c3fa5d3e18804","project":"openstack/releases","branch":"master","topic":"auto-releases","hashtags":[],"change_id":"I6ec9fa77baab58df93bdadc0ac3c3fa5d3e18804","subject":"Add script to semi-automate bulk releases","status":"MERGED","created":"2020-09-11 19:33:37.000000000","updated":"2020-09-21 07:43:17.000000000","submitted":"2020-09-21 07:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751409-1600674197744-515a9e8b","meta_rev_id":"43adfb0f47aa0207d7c9642bc00267e2463bab96","_number":751409,"virtual_id_number":751409,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 07:43:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-09-14 09:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-09-14 11:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-16 10:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-09-21 07:30:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 11:51:07.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-16 10:59:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-21 07:30:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-21 07:43:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30b6febb32c4bdfb2e5626d50817dede7f85a064","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:33:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b697afbb4abf9cbbfd8a1cfd234e827b81d8fb0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:49:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c830f4ecdd98a40cab7c907b8ec00472f4909a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:27:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3d4f86da4024c90baf6d1878fb4f880 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73266c9a1b5247c28e48427c89315455 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f036bc4603044d76bd9d63b5667fb362 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c28415ea9a1742fc988991df55f1eae9 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e2fede462414eee93024ebb0b486bf1 : SUCCESS in 6m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/22d58f78a51744fa96065e62f553e80c : SUCCESS in 31m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56cd8a10fa0640cca2c5baa834653114 : SUCCESS in 5m 24s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e05d4c92064e4a60a557a6799f97bda4 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6d95a7a99479fba695e5933e3da70ba5d1b59657","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-14 09:05:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"88bff3403e2ad66e681aee234a089ca1595e55fa","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-14 11:51:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24cff089ec75429df940bdb1c965e5f0e9152043","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 10:59:00.000000000","message":"Patch Set 2: Code-Review+2\n\nI guess we need to use this script during milestone-3 as it was the case last week, so maybe it could be worth to add some related note in the process\u0027s documentation [1], isn\u0027t?\n\nIf yes it could be done with a follow up patch.\n\n[1] https://opendev.org/openstack/releases/src/branch/master/doc/source/reference/process.rst","accounts_in_message":[],"_revision_number":2},{"id":"35308bec0e5407e31ef8c7e3f40f34ffad719b06","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 07:30:56.000000000","message":"Patch Set 2: Workflow+1\n\nto be done in a following patch if it works","accounts_in_message":[],"_revision_number":2},{"id":"727255e558d3d27f4c6aea338b49f649ef862510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:31:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9ad44f4bafefb413b2325e7ae3c8dd8270b4e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:43:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e3e14d1d6a5c30ec22ce6bfa5447b5529e92d1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:43:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7dc410108b34ffb85af28bc7371d785 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2776c4224b48c083c1fc234c84ac5a : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5422b191779c4498bb36b02b01ee38b4 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c0d8658368485ca385a5fc3cc1f187 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fc88e37e0d14b78b59548e3750c38d1 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f152ac34232641f69bcb2e967a053b44 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"355cd7debdd6316ee29ccb8b6c1d3aa1aebd2f19","revisions":{"0c92a45ee8d67fc90259cefcdd13c6aa37a5e0b4":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:33:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/09/751409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/751409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/751409/1"}}},"commit":{"parents":[{"commit":"5617dcc77f91e20e92e4f02ff25cebda0f203f32","subject":"Merge \"Final Oslo releases for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5617dcc77f91e20e92e4f02ff25cebda0f203f32"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-15 19:34:46.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 19:33:31.000000000","tz":-300},"subject":"Add script to semi-automate bulk releases","message":"Add script to semi-automate bulk releases\n\nThe release team has a few cases now where we need to propose release\npatches for a large set of deliverables. Due to differences in semver\nchoice based on actually merged commits, along with other decisions that\nneed to be made while doing these, this process can\u0027t be completely\nautomated. But this adds a script that will automate the majority of the\nprocess to simplify it as much as possible.\n\nChange-Id: I6ec9fa77baab58df93bdadc0ac3c3fa5d3e18804\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c92a45ee8d67fc90259cefcdd13c6aa37a5e0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c92a45ee8d67fc90259cefcdd13c6aa37a5e0b4"}]},"branch":"refs/heads/master"},"355cd7debdd6316ee29ccb8b6c1d3aa1aebd2f19":{"kind":"REWORK","_number":2,"created":"2020-09-11 19:49:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/09/751409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/751409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/751409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/751409/2"}}},"commit":{"parents":[{"commit":"5617dcc77f91e20e92e4f02ff25cebda0f203f32","subject":"Merge \"Final Oslo releases for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5617dcc77f91e20e92e4f02ff25cebda0f203f32"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-15 19:34:46.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-11 19:48:55.000000000","tz":-300},"subject":"Add script to semi-automate bulk releases","message":"Add script to semi-automate bulk releases\n\nThe release team has a few cases now where we need to propose release\npatches for a large set of deliverables. Due to differences in semver\nchoice based on actually merged commits, along with other decisions that\nneed to be made while doing these, this process can\u0027t be completely\nautomated. But this adds a script that will automate the majority of the\nprocess to simplify it as much as possible.\n\nChange-Id: I6ec9fa77baab58df93bdadc0ac3c3fa5d3e18804\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/355cd7debdd6316ee29ccb8b6c1d3aa1aebd2f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/355cd7debdd6316ee29ccb8b6c1d3aa1aebd2f19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
