)]}'
{"id":"openstack%2Freleases~752972","triplet_id":"openstack%2Freleases~master~I451f68ab712a250ec9867f3455c66105b3205853","project":"openstack/releases","branch":"master","topic":"victoria-rc1","hashtags":[],"change_id":"I451f68ab712a250ec9867f3455c66105b3205853","subject":"Proposing victoria RC1 for adjutant","status":"MERGED","created":"2020-09-21 14:47:59.000000000","updated":"2020-09-25 08:32:38.000000000","submitted":"2020-09-25 08:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752972-1601022758095-e5a30d7c","meta_rev_id":"110d6346d9cf8d4802fbb94c5ac36f5b174bc6df","_number":752972,"virtual_id_number":752972,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 08:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"all":[{"value":1,"date":"2020-09-22 10:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-24 21:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":2,"date":"2020-09-25 08:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-09-25 08:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-09-25 08:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-24 21:16:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 10:04:11.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-24 21:15:14.000000000","updated_by":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"reviewer":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"state":"REVIEWER"},{"updated":"2020-09-25 08:17:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-25 08:32:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb2971a485567116106da95f2e8bfc13e6a6ef87","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 14:47:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bedf9d0c5ef711bf535b395cef3a00b3ad9de00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:52:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ec8811222a4cfdb91bcf1526c2f006 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c141b5a186aa492588f7ef54d5cc60b7 : SUCCESS in 6m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3b29d0b9a729434a967320c90780b2b2 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"10ef59d075bdec853d0678875c5c0a579dd84790","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 15:57:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdd6ed8feda65b0c714edcc27b20d777b6427ef3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 20:39:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aebad36918c0e47ef201a7dd1ae91796799605fd","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 10:04:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"85ac81f6aade964b0ff10c6248e40a6e1ceca6bf","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 10:25:22.000000000","message":"Patch Set 1:\n\nLGTM for bump and release-model point of view; waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"e6be7f7265b0bd6028f165b8ab52fe730245aefe","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-09-24 21:15:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"abb8b35036e26c3b7f358b2e7ff14ab410c5683c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 21:16:54.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2ddb508686cb99ff89b61a66cb4e72c6762dca4a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-25 08:14:06.000000000","message":"Patch Set 1: Code-Review+2\n\nDeliverables following the RC model use major version bumps at each release, so it\u0027s normal that we bump to 1.0.0 here","accounts_in_message":[],"_revision_number":1},{"id":"b16f24c491cc1831d5e9181a39a1f833359a5c93","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-25 08:17:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e Deliverables following the RC model use major version bumps at each\n \u003e release, so it\u0027s normal that we bump to 1.0.0 here\n\nAck thanks for the heads up","accounts_in_message":[],"_revision_number":1},{"id":"f409913c98cfceb92993c4fbf88138bb2c4b021b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:17:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d3dc9c3df93c7f390d4abec617a6b81c4123410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:32:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5781cf5730a84582893cb3b94807f8c2 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a6107a612643088827b0c197b793d0 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94e32a2244884e0fb4a9f87cc018366d : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8247cfc67d20e88b01b77a96786c688917c9fdfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:32:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34ce745b06265cb7461b537a62402fb64d632317","revisions":{"34ce745b06265cb7461b537a62402fb64d632317":{"kind":"REWORK","_number":1,"created":"2020-09-21 14:47:59.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/72/752972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/752972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/752972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/752972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/752972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/752972/1"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 14:47:53.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 14:47:53.000000000","tz":120},"subject":"Proposing victoria RC1 for adjutant","message":"Proposing victoria RC1 for adjutant\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for adjutant.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: I451f68ab712a250ec9867f3455c66105b3205853\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34ce745b06265cb7461b537a62402fb64d632317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34ce745b06265cb7461b537a62402fb64d632317"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
