)]}'
{"id":"openstack%2Freleases~752997","triplet_id":"openstack%2Freleases~master~I9fef5ed3ed06e1ae66d1b7c554c9ffbf97f3f53e","project":"openstack/releases","branch":"master","topic":"victoria-rc1","hashtags":[],"change_id":"I9fef5ed3ed06e1ae66d1b7c554c9ffbf97f3f53e","subject":"Proposing victoria RC1 for kolla","status":"ABANDONED","created":"2020-09-21 15:11:37.000000000","updated":"2020-09-25 08:09:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"722007b0a5572f3f3b96ddad0dbc0f0edcaf2eb1","_number":752997,"virtual_id_number":752997,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-21 17:19:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-22 11:04:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":-1,"date":"2020-09-21 16:08:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 16:08:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-21 17:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 11:04:17.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"}],"messages":[{"id":"506ca4a8120d0da0fedb6b1593d01a2e01139d35","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 15:11:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7f7c9d7966f9b721e9a6bdfc6b5cb863acaf0f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 16:08:09.000000000","message":"Patch Set 1: Code-Review-1\n\nKolla is cycle-trailing","accounts_in_message":[],"_revision_number":1},{"id":"194e06bed7fedcedf8e1d1ed5e9e2f6523d474a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:19:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f520e270f6e2440792f4c312f4aa202c : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6674379555a44cd7839f2ef90fad25e6 : SUCCESS in 4m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/db9bf72e75d04652a60ed016e35ca5d3 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e0844954409057f94b0b1dd71a2fe68897741d5f","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 11:04:17.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM for bump and release-model point of view; waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"8ad4343fba32144546964afb9170f95b54f4f692","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-25 08:09:07.000000000","message":"Abandoned\n\nYou are right Mark -- this was caught in error. Sorry for the confusion.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cf88510b93235d460819dc729c217b09053ae0e","revisions":{"7cf88510b93235d460819dc729c217b09053ae0e":{"kind":"REWORK","_number":1,"created":"2020-09-21 15:11:37.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/97/752997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/752997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/752997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/752997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/752997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/752997/1"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:11:31.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:11:31.000000000","tz":120},"subject":"Proposing victoria RC1 for kolla","message":"Proposing victoria RC1 for kolla\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for kolla.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: I9fef5ed3ed06e1ae66d1b7c554c9ffbf97f3f53e\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cf88510b93235d460819dc729c217b09053ae0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cf88510b93235d460819dc729c217b09053ae0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
