)]}'
{"id":"openstack%2Freleases~753004","triplet_id":"openstack%2Freleases~master~I8b5d4bc6d46ed33c28a4a98635ddabfd5475edfe","project":"openstack/releases","branch":"master","topic":"victoria-rc1","hashtags":[],"change_id":"I8b5d4bc6d46ed33c28a4a98635ddabfd5475edfe","subject":"Proposing victoria RC1 for masakari-dashboard","status":"MERGED","created":"2020-09-21 15:16:06.000000000","updated":"2020-09-28 15:37:42.000000000","submitted":"2020-09-28 15:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753004-1601307462558-1561717f","meta_rev_id":"884de985dd8904807f6c13f4bb250e56a891f11b","_number":753004,"virtual_id_number":753004,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 15:37:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-28 15:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-25 08:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-25 08:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-09-23 12:59:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-28 15:22:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 15:20:44.000000000","updated_by":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"reviewer":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"state":"REVIEWER"},{"updated":"2020-09-22 09:53:56.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-23 12:59:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-25 08:21:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-28 15:22:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-28 15:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf12f6743237b937eff25800fe47614a177437b4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 15:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76bad61c8e81675c7228dcfd9694979846e1e423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:46:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece3a027028d433f97aadf96ea11dc94 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25f0490d37b642788228cbe1da7e4e1b : SUCCESS in 7m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/56b050b9407e453bb73614dd85e582e0 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"50ab7c1d529d3c1a9df8a25080409946ac88f4ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 17:49:46.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"0287ae54c7c9f46d5cafed6a83a07c1aff422db7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-22 07:23:36.000000000","message":"Patch Set 1:\n\nregarding masakari, there are a few pending fixes I\u0027d like to get in RC1 (and final realease for sure) to avoid unnecessary commotion but they are not release-critical","accounts_in_message":[],"_revision_number":1},{"id":"49578d5128488d3a6e7122d5526edae0b20a7983","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 09:53:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ca6daf28db2f378609ff793516c070c0bc176fe","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 10:26:32.000000000","message":"Patch Set 1:\n\nLGTM for bump and release-model point of view; waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"bced77332972d9e6e15e849f48402cf3bfc6bae5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-23 12:59:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"159e6358988e7913411412cfca3f91de13067374","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-23 12:59:37.000000000","message":"Patch Set 2: Code-Review+1\n\nall in","accounts_in_message":[],"_revision_number":2},{"id":"e7999aafe198f40608d82daf9e2c435ad4ec8a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 13:24:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65008ac8e63b4b76940f7488089c9370 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/934614b138cc42f8a80ae3dd4a19fbd7 : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9a14352085e345edbe68618c8278d892 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"be2a5555d5489ce717f01f62214971a513f558ad","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-25 08:18:30.000000000","message":"Patch Set 2: Code-Review+2\n\nHappy to consider Radosław as the release liaison as Sampath is MIA","accounts_in_message":[],"_revision_number":2},{"id":"05b5aa99e5e1adccacca0c56bcc51358fbb1cb34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-25 08:21:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93f842ff404f6fd1a186399eb9186549999023c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 15:22:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8afaa746cee48a97ad547c9800261f4ef864b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:24:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb6c1354167345ac20b5efa3168ce5f1079c1974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:37:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7fea902eb9dd66179ffec6422f710e4e4618a2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:37:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d27ad1c451540d997bc58ded5302185 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b09e52c28a14604a88762e72962d467 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4283be8bbb05468c91ca9f98ccca63de : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da90d772cb711958b1880982dd87d37977de4aab","revisions":{"ff65ea8de6d85f78172377efed269ea4969a3181":{"kind":"REWORK","_number":1,"created":"2020-09-21 15:16:06.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/753004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/753004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/753004/1"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:16:00.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:16:00.000000000","tz":120},"subject":"Proposing victoria RC1 for masakari-dashboard","message":"Proposing victoria RC1 for masakari-dashboard\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for masakari-dashboard.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: I8b5d4bc6d46ed33c28a4a98635ddabfd5475edfe\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff65ea8de6d85f78172377efed269ea4969a3181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff65ea8de6d85f78172377efed269ea4969a3181"}]},"branch":"refs/heads/master"},"da90d772cb711958b1880982dd87d37977de4aab":{"kind":"REWORK","_number":2,"created":"2020-09-23 12:59:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/753004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/753004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/753004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/753004/2"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:16:00.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-23 12:59:30.000000000","tz":0},"subject":"Proposing victoria RC1 for masakari-dashboard","message":"Proposing victoria RC1 for masakari-dashboard\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for masakari-dashboard.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: I8b5d4bc6d46ed33c28a4a98635ddabfd5475edfe\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da90d772cb711958b1880982dd87d37977de4aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da90d772cb711958b1880982dd87d37977de4aab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
