)]}'
{"id":"openstack%2Freleases~753014","triplet_id":"openstack%2Freleases~master~Iae8622e48865a30352dc0ab51f22c7805405e8da","project":"openstack/releases","branch":"master","topic":"victoria-rc1","hashtags":[],"change_id":"Iae8622e48865a30352dc0ab51f22c7805405e8da","subject":"Proposing victoria RC1 for monasca-notification","status":"MERGED","created":"2020-09-21 15:19:35.000000000","updated":"2020-09-28 15:33:18.000000000","submitted":"2020-09-28 15:33:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753014-1601307198881-276e733d","meta_rev_id":"4082174944d018704ad1d7a55b7764d0fca27f41","_number":753014,"virtual_id_number":753014,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 15:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-28 15:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-22 10:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-21 18:21:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-28 15:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-21 15:20:10.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 15:20:10.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-09-21 18:21:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-22 10:49:59.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-28 15:16:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-28 15:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec19b4497dbac25ccb1f9839dce46b830b26687f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 15:19:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e08f5c409369980a588c1b20bd77dd145b60747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6980d049d73d4a07a86ed227d0afc91c : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebb9de6f42a94797824b37c0f1ee04c8 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/34ea49a964724a76a4dfc6ffedb48816 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e10a22d29b5183dc6014025b32fec5474aee3b8a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 18:21:14.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"c4b1ec3a6c5c89b821c2ae9c8040f24d3402c6ff","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 10:49:59.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM for bump and release-model point of view; waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"5818e684df354298dfa5258929203645bcb25342","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 15:16:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPast deadline, commit hash is still current HEAD.","accounts_in_message":[],"_revision_number":1},{"id":"7ca690a8ec7922392fb3d5de14cb9a86c3511974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:16:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e5ee563a179d53f33030c879a4201253d9d74f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:33:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"71f538dc9b8fde6f4bbc57e3e826129f59c8b25d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:33:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49e13839f1a4031a7c7fe13acb0d039 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f455ef2f2414dcba3faeefa12549895 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0117cbb2279e45e1b34a223cfff9d9d0 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cf9bfcd6c6e80fc17b6727b1f81f03441191175","revisions":{"2cf9bfcd6c6e80fc17b6727b1f81f03441191175":{"kind":"REWORK","_number":1,"created":"2020-09-21 15:19:35.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/14/753014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/753014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/753014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/753014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/753014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/753014/1"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:19:29.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:19:29.000000000","tz":120},"subject":"Proposing victoria RC1 for monasca-notification","message":"Proposing victoria RC1 for monasca-notification\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for monasca-notification.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: Iae8622e48865a30352dc0ab51f22c7805405e8da\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cf9bfcd6c6e80fc17b6727b1f81f03441191175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cf9bfcd6c6e80fc17b6727b1f81f03441191175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
