)]}'
{"id":"openstack%2Freleases~753019","triplet_id":"openstack%2Freleases~master~Ic47e7adba3ba50558728478dd69c038e3d0bb86a","project":"openstack/releases","branch":"master","topic":"victoria-rc1","hashtags":[],"change_id":"Ic47e7adba3ba50558728478dd69c038e3d0bb86a","subject":"Proposing victoria RC1 for monasca-ui","status":"MERGED","created":"2020-09-21 15:20:57.000000000","updated":"2020-09-28 15:33:17.000000000","submitted":"2020-09-28 15:33:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753019-1601307197067-e82f702d","meta_rev_id":"012d0d25595f65daca5be471e7645c38482ac3a5","_number":753019,"virtual_id_number":753019,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 15:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-28 15:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-22 10:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-28 15:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-28 15:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-21 16:26:02.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 16:26:02.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-09-22 10:42:53.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-28 15:15:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-28 15:23:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-28 15:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64d2736b93c6614b9592c9a22c24eef59a1a9cd6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-21 15:20:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b5657c6210676096f90efe6ca82dc357959905d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 18:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0e04908aef4be998e970e77cf44c9c : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24aa85e9882c4a0a8e9802a10a756ad4 : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/eff9f88fae704b8b883f703a74d5106e : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"275aaf32049ce0260dfcac3e7d149a83f901a048","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 18:27:34.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"41ccc8a5ee068495c2a6c526d36a6e5c57d33432","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 10:42:53.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM for bump and release-model point of view; waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"558f04c17370886a8ceba1c681cfe4b61749e1cf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-28 15:15:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPast deadline, commit hash is still current HEAD.","accounts_in_message":[],"_revision_number":1},{"id":"79cd475f4a7d4dcf73858f1409e1f20bd85152d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:15:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d1ce163b66ad7130733e24bbaf10df0d358acf6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-28 15:23:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ddbac886dd3d3600ba93a4f4ed0451434b773c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:33:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9b827be9984cf78b304e056a7b700c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5717e6ac5e1481680621fc7ee722071 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7195cd8543a4a5d96f72e195359ca41 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"efa547c8fe0f00da10c047f2f7dc093ec317e8dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:33:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c97c63e9c1a9dbae67f8201e880b61fafec5b46","revisions":{"6c97c63e9c1a9dbae67f8201e880b61fafec5b46":{"kind":"REWORK","_number":1,"created":"2020-09-21 15:20:57.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/19/753019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/753019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/753019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/753019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/753019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/753019/1"}}},"commit":{"parents":[{"commit":"01ed0b46a15562b883b855ce98a6de524240d766","subject":"Merge \"Release Patrole 0.10.0 for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01ed0b46a15562b883b855ce98a6de524240d766"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:20:51.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-09-21 15:20:51.000000000","tz":120},"subject":"Proposing victoria RC1 for monasca-ui","message":"Proposing victoria RC1 for monasca-ui\n\nAs a courtesy for project teams, on RC1 week the release management\nteam proposes a RC1 release for all cycle-with-rc deliverables,\nbased on HEAD. That RC1 will double as the branch point for the\nstable/victoria branch for monasca-ui.\n\nRelease liaison or PTL should +1 this proposal if they are happy\nwith it, amend this proposal (for example with a more recent SHA)\nto match their needs, or -1 to block it until release-critical\nissues are all fixed.\n\nChange-Id: Ic47e7adba3ba50558728478dd69c038e3d0bb86a\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c97c63e9c1a9dbae67f8201e880b61fafec5b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c97c63e9c1a9dbae67f8201e880b61fafec5b46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
