)]}'
{"id":"openstack%2Freleases~753952","triplet_id":"openstack%2Freleases~master~I9b14f391ca8deb25f7f38c0426f17a3674263674","project":"openstack/releases","branch":"master","topic":"release-ovn-octavia-provider-for-victoria","hashtags":[],"change_id":"I9b14f391ca8deb25f7f38c0426f17a3674263674","subject":"Release ovn-octavia-provider 0.2.1 and cut stable/victoria branch","status":"MERGED","created":"2020-09-24 06:33:39.000000000","updated":"2020-09-24 16:20:04.000000000","submitted":"2020-09-24 16:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753952-1600964404220-0e5654ba","meta_rev_id":"65f7d7f233cff4dcffe13ae73bd18ef9982aa2f5","_number":753952,"virtual_id_number":753952,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 16:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-24 16:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-09-24 13:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-24 16:07:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-24 12:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 07:39:30.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-24 11:20:24.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-24 13:01:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-24 16:07:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-24 16:20:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac6155e55966ecd8be454b942630aa30e967b07e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 06:33:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7a8623e757c440a36ad86efbc742bfb49bd1998","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:34:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"6028617c0f51613ddf3a9266ea7cd1ce931b2bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e94e9950f9e4fbbb9efeca1719e0de0 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7389ed1ddbc549528f0980f08b8b3286 : SUCCESS in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1be13180b6b47568b720e05ff707531 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"20766b86b66af6edd295cb2943b6f90b2c564eb3","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-24 07:39:30.000000000","message":"Patch Set 1: Code-Review-1\n\nShould be a minor bump since requirements ar updated","accounts_in_message":[],"_revision_number":1},{"id":"00110b829071e7473b9d0ac4db6fb3a3960b6d28","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 08:04:50.000000000","message":"Patch Set 1:\n\n\u003e Should be a minor bump since requirements ar updated\n\nAgreed\n\noslo.messaging and SQLAlchemy have been bumped so it should be a minor version at least \n\nAlso this project doesn\u0027t follow the stable branch policy [1] so requirements updates are not an issue.\n\n[1] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html","accounts_in_message":[],"_revision_number":1},{"id":"defe7547131a4426a5aec99a03b7479e8dcb1c29","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 12:23:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"25eb6fa0d69474fdd3ba05914b18a6d1a46c782e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:25:45.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3a05c301a1e40bb553fad63b3ac1cb8d2b47b9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:43:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdfeac2173245f6a256c588685ee46b : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39ef0a650f56401ba5d5dd158db943ae : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/13b14cbd9707499797b9d5625528d975 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"81bd48b7138ad1af24228ebc709d43962f6d2753","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 13:01:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca8ee2caf9032386e15f7a4a6d9ab12ee4cac265","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-24 16:07:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"571e3ef23863f4f33261d7f514e84841fbd88ba7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:07:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42726b70b9323e35721be06fb00fd4502d9238c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:20:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9482b133a24e7aafc3442b6d20a8b6 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e82f0bdc706497d83c851b1b7837262 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/352382bc84ef45e1ada7ef163e7daf25 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8a43bd30e2362b26b16883a55d024dceaea62108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:20:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7fcdb7ccb1bdc18b0806581301897fe636041c5","revisions":{"c8aa993a417d4ee1a192f55beac608d4a85474c1":{"kind":"REWORK","_number":1,"created":"2020-09-24 06:33:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/753952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/753952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/753952/1"}}},"commit":{"parents":[{"commit":"3ca70158e344f7b0234699d93d0c7ad960c40eeb","subject":"Merge \"Proposing victoria RC1 for zun\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ca70158e344f7b0234699d93d0c7ad960c40eeb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 06:30:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 06:33:35.000000000","tz":120},"subject":"Release ovn-octavia-provider 0.2.1 and cut stable/victoria branch","message":"Release ovn-octavia-provider 0.2.1 and cut stable/victoria branch\n\nChange-Id: I9b14f391ca8deb25f7f38c0426f17a3674263674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8aa993a417d4ee1a192f55beac608d4a85474c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8aa993a417d4ee1a192f55beac608d4a85474c1"}]},"branch":"refs/heads/master"},"c7fcdb7ccb1bdc18b0806581301897fe636041c5":{"kind":"REWORK","_number":2,"created":"2020-09-24 12:23:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/753952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/753952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/753952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/753952/2"}}},"commit":{"parents":[{"commit":"3ca70158e344f7b0234699d93d0c7ad960c40eeb","subject":"Merge \"Proposing victoria RC1 for zun\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ca70158e344f7b0234699d93d0c7ad960c40eeb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 06:30:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 12:23:51.000000000","tz":0},"subject":"Release ovn-octavia-provider 0.2.1 and cut stable/victoria branch","message":"Release ovn-octavia-provider 0.2.1 and cut stable/victoria branch\n\nChange-Id: I9b14f391ca8deb25f7f38c0426f17a3674263674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7fcdb7ccb1bdc18b0806581301897fe636041c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7fcdb7ccb1bdc18b0806581301897fe636041c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
