)]}'
{"id":"openstack%2Freleases~754059","triplet_id":"openstack%2Freleases~master~I38378e3cae64d6b7c7ed22030e8f43c112d865ca","project":"openstack/releases","branch":"master","topic":"vitrage-v-highlights","hashtags":[],"change_id":"I38378e3cae64d6b7c7ed22030e8f43c112d865ca","subject":"Add cycle highlights for Vitrage","status":"MERGED","created":"2020-09-24 10:09:20.000000000","updated":"2020-09-24 16:22:27.000000000","submitted":"2020-09-24 16:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754059-1600964547431-6ef20142","meta_rev_id":"ef8538f26ab62ed8190a1f15cae55489de789024","_number":754059,"virtual_id_number":754059,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-09-24 12:20:44.000000000","_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 16:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-09-24 16:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":1,"date":"2020-09-24 14:10:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-09-24 12:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-24 13:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-09-24 16:07:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-09-24 13:04:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 12:43:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-24 13:03:28.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-24 14:10:46.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-24 16:07:55.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-09-24 16:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88b46bd7a9a3a1838858fb4cdfe7351d4aaf9169","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-24 10:09:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b279ac826eff9a7424523a4bef552febf4c8a81","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-24 10:12:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d129c45fa1b4c06b0fd22ae96ae51439e477eef4","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-24 10:20:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2fb24873500f125877339da311eef81d8256fa58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:38:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01131d8f69974d0dbabeb8a28cace309 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d90bcbba0c6c4770a6e4e3341d72b6ab : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b1bb2b4b32514c93ac1471798f971f19 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"736b6be7bdaae15d4e20069767ce277f23b0058a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-24 10:40:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"526de0c7a5beb71fa7563aae634cdf3ed61eda54","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:41:35.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"8f744aa519a0ab4042e5701539d1d28cc66c69f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 11:52:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"351e982e9a21697825141c178c30cba23c8262bb","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-24 12:19:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e700385fcc393af8e8f9cf9e629f1a3671819187","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-24 12:20:44.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nDone.","accounts_in_message":[],"_revision_number":4},{"id":"abb18a8bf8b194acef71d6ee3544c679e5b76bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:32:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d997402b08a40bebd5e14545988bd0e : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bbb35e3566f43fb8b8f0fa687796a15 : SUCCESS in 4m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e560d419f4941209d365ef55093aa56 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7c453ceb669dcf7e4fc37e2ecaba90716b42f4a9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 12:43:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6d0ae5e87c1d1c72650a031e3dd46b0dfeccb9d6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-24 13:03:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9ec25791fe586ae1a0dfa9d55f96734df29db6eb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:04:52.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"221e0a8d14a80dcadfaa9933344f3903176b0ab9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-24 14:10:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d922ef93b5d297d0871d00dc30f943c79df162ac","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-09-24 16:07:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e9ee7d0613ca9ea5bea2211ac62a9e0d55d0e5f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:08:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"50164ec6462f176ae257bc3196dbb0aa05908eba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:22:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fced37fa81ba4195949fdf7bb12f31ec : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0017162f4f4b4fce9eed0f43da7dc63c : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e17f31dacdfb44279998fac6c097ee69 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"9e1c489f3daa72f563a2529bb61ce05bc8d84e45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:22:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"adc81df39abe405009513fd7891423fbc1539922","revisions":{"6df9b70402cfa0e72741273e08f4c9e00863bae7":{"kind":"REWORK","_number":1,"created":"2020-09-24 10:09:20.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/59/754059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/754059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/754059/1"}}},"commit":{"parents":[{"commit":"cfbfb01937b2e19c1efff7c29ebce98d0f99772c","subject":"Merge \"Proposing victoria RC1 for glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfbfb01937b2e19c1efff7c29ebce98d0f99772c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:08:56.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:08:56.000000000","tz":480},"subject":"Add cycle highlights for Vitrage","message":"Add cycle highlights for Vitrage\n\nChange-Id: I38378e3cae64d6b7c7ed22030e8f43c112d865ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6df9b70402cfa0e72741273e08f4c9e00863bae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6df9b70402cfa0e72741273e08f4c9e00863bae7"}]},"branch":"refs/heads/master"},"faeb4d4e08d5a5e96b86b4caedcd3303ad726ff1":{"kind":"NO_CHANGE","_number":2,"created":"2020-09-24 10:12:22.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/59/754059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/754059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/754059/2"}}},"commit":{"parents":[{"commit":"cfbfb01937b2e19c1efff7c29ebce98d0f99772c","subject":"Merge \"Proposing victoria RC1 for glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfbfb01937b2e19c1efff7c29ebce98d0f99772c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:08:56.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:11:34.000000000","tz":480},"subject":"Add cycle highlights for Vitrage","message":"Add cycle highlights for Vitrage\n\nChange-Id: I38378e3cae64d6b7c7ed22030e8f43c112d865ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faeb4d4e08d5a5e96b86b4caedcd3303ad726ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faeb4d4e08d5a5e96b86b4caedcd3303ad726ff1"}]},"branch":"refs/heads/master"},"0cbf247691c66523387ed56ab2bff07101265c27":{"kind":"REWORK","_number":3,"created":"2020-09-24 10:20:52.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/59/754059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/754059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/754059/3"}}},"commit":{"parents":[{"commit":"cfbfb01937b2e19c1efff7c29ebce98d0f99772c","subject":"Merge \"Proposing victoria RC1 for glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfbfb01937b2e19c1efff7c29ebce98d0f99772c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:08:56.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:17:26.000000000","tz":480},"subject":"Add cycle highlights for Vitrage","message":"Add cycle highlights for Vitrage\n\nChange-Id: I38378e3cae64d6b7c7ed22030e8f43c112d865ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbf247691c66523387ed56ab2bff07101265c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbf247691c66523387ed56ab2bff07101265c27"}]},"branch":"refs/heads/master"},"adc81df39abe405009513fd7891423fbc1539922":{"kind":"REWORK","_number":4,"created":"2020-09-24 12:19:22.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/59/754059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/754059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/754059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/754059/4"}}},"commit":{"parents":[{"commit":"cfbfb01937b2e19c1efff7c29ebce98d0f99772c","subject":"Merge \"Proposing victoria RC1 for glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cfbfb01937b2e19c1efff7c29ebce98d0f99772c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 10:08:56.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-24 12:16:14.000000000","tz":480},"subject":"Add cycle highlights for Vitrage","message":"Add cycle highlights for Vitrage\n\nChange-Id: I38378e3cae64d6b7c7ed22030e8f43c112d865ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adc81df39abe405009513fd7891423fbc1539922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adc81df39abe405009513fd7891423fbc1539922"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
