)]}'
{"id":"openstack%2Freleases~755025","triplet_id":"openstack%2Freleases~master~I31bd4e1b441f92cd592520df8d5ef482bdfe530c","project":"openstack/releases","branch":"master","topic":"victoria-c-a","hashtags":[],"change_id":"I31bd4e1b441f92cd592520df8d5ef482bdfe530c","subject":"Release freezer-tempest-plugin for victoria","status":"MERGED","created":"2020-09-29 14:20:47.000000000","updated":"2020-10-02 12:10:35.000000000","submitted":"2020-10-02 12:10:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755025-1601640635675-5cec80e2","meta_rev_id":"e65f27ad80d795783cac78397b2be3694ef1ca8d","_number":755025,"virtual_id_number":755025,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 12:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":2,"date":"2020-10-02 11:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":1,"date":"2020-10-02 11:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"date":"2020-09-29 14:27:09.000000000","_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-09-29 14:27:09.000000000","_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 14:27:09.000000000","updated_by":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2020-09-29 14:27:09.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-10-02 11:56:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-02 12:10:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7852cae85dcbb4c38750ea768b19d3983c2517e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-29 14:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09faea35b31756e06e1c4aee6dd81a39e6ba0098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 14:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef9aedd5ed84899a43b86987cdd063c : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/001d1bccbe8b4dcd924859488b5af29c : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/23a83c2a1cee463ea02f830dfbcb1cd7 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b9d2391b630217104af72a7fa0f97534ab01ce11","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 20:51:25.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, Waiting for PTL\u0027s approval","accounts_in_message":[],"_revision_number":1},{"id":"4611dda3c22d74e0a908483cbc7bc16912b55942","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 11:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nDue date reached","accounts_in_message":[],"_revision_number":1},{"id":"18f47880ffcd6f42f2da4ec67f6227265023725b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:56:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2d3f129606ae4ce4244ae47e5bab4a18d493672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:10:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c859b4e7fb5f433a96eefac58b844828 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7696d6045a5a4f3e814412662b5dc3db : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2610d338177243b2b1a2c7b360471fec : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9eb31ea2cff1729aee5692d5317cb8d815bafd95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:10:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66e8eee86ba610558908a3e823ac0a5fa64b20e3","revisions":{"66e8eee86ba610558908a3e823ac0a5fa64b20e3":{"kind":"REWORK","_number":1,"created":"2020-09-29 14:20:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/25/755025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/755025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/755025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/755025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/755025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/755025/1"}}},"commit":{"parents":[{"commit":"84be2d68540ebc0ecfcc12e3f1e9018894350628","subject":"Merge \"puppet: make stable/ocata and stable/pike eol\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84be2d68540ebc0ecfcc12e3f1e9018894350628"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-29 14:20:45.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-29 14:20:45.000000000","tz":-300},"subject":"Release freezer-tempest-plugin for victoria","message":"Release freezer-tempest-plugin for victoria\n\nThis is the automatic release for freezer-tempest-plugin.\nPlease +1 if this is ready to go, or -1 if the team needs a little more\ntime. If so, please update with the new commit hash to use when ready.\n\nIf no response, this will be approved by October 2.\n\nChange-Id: I31bd4e1b441f92cd592520df8d5ef482bdfe530c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66e8eee86ba610558908a3e823ac0a5fa64b20e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66e8eee86ba610558908a3e823ac0a5fa64b20e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
