)]}'
{"id":"openstack%2Freleases~755362","triplet_id":"openstack%2Freleases~master~Ib1fe5f74b408ac489b12c4ef003860ce7aee1532","project":"openstack/releases","branch":"master","topic":"release_on_mailing_list","hashtags":[],"change_id":"Ib1fe5f74b408ac489b12c4ef003860ce7aee1532","subject":"Adding a new tools to search topic on the ML","status":"MERGED","created":"2020-09-30 17:59:49.000000000","updated":"2020-10-19 16:25:27.000000000","submitted":"2020-10-19 16:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755362-1603124727973-795c03e7","meta_rev_id":"8bde2f8f8155c48d5f488f33fc41b5e9b5649bbe","_number":755362,"virtual_id_number":755362,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"date":"2020-10-02 13:59:38.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 16:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-19 16:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-07 15:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-05 08:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-05 08:51:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-19 16:08:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 10:21:25.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-10-05 08:42:33.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-10-05 08:51:44.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-07 15:01:12.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-19 16:08:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-19 16:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d2974e2e26e2899e7d1207c23c34a829a3b1169","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-30 17:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34fb4604b21fe2785a39c476673ed8a76ca71be4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-30 18:01:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"642587465b6c67124daa46e7900523559c15f487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:48:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ce410c24744d73a54a149de1ac1b34 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b576f2997c04079a89d3dee08e9f6b6 : TIMED_OUT in 40m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5854a6d7ec9447c1a97bef2d708af2aa : TIMED_OUT in 41m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c336fe037e641ceba0569e0b9298cb5 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb3eb5bdd3de436c943670ea9e8d3f8c : SUCCESS in 8m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/549d08a7503749b896060550250eba0d : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef9f1959d65647d3901b3ce320e642c4 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"22e3f92d354df01cf3b18493fb3f8d1876ba9689","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-30 19:52:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8259ffe24957bd3a1883fe144eba2694024f0676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:40:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fb82d9594b4450a893dff4b6bee508 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f89a3ea39734e5f83f8dbf1a81690d5 : TIMED_OUT in 41m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5be0e20ea0bc414b858423159147b90d : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6ae38323ca48f094b451f1e0bd40a5 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c672d05a7e0f43799806d11729218143 : SUCCESS in 4m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/21d726ab8c4b453aa3444e4892cbf4aa : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88e5115a8129447a930e8e920daa086a : TIMED_OUT in 41m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9fe893ed06d6fdc67181e6cf6fbc210352e75de3","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-01 05:52:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4aca90d46e3ae2c6fa02079bcec12114b80995b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 05:59:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be7d50a94b24cb7b6c6d8a711a93486 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042538a787b9486ba0839ac03b3e0634 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bc95f51036a434ebb75b8be4684a5dc : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785f5aedc16a4cc38dab64ca1c55e153 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cb3055e85f7408ab0695c4bc50f7845 : SUCCESS in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f99a15a94968424bb72b8020e23187f0 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70e5e142758a4577a3a6a5ca3f407eb6 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"bab8ccef0d29581319a2f2421803725fbc1749ea","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-01 06:11:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3a78041a085d866a0cd9863a3d8b540f9b4fd4e5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-10-01 08:06:41.000000000","message":"Patch Set 2: Code-Review+1\n\nNice tool!, i\u0027ve successfully tested with other ML as RDO ones just by modifying BASE_URL so it\u0027d be nice to have a way to override it via CLI parameter.","accounts_in_message":[],"_revision_number":2},{"id":"50f37b3d473be6a6029a68864b3505cb3320b05e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 12:29:52.000000000","message":"Patch Set 2:\n\n\u003e Nice tool!, i\u0027ve successfully tested with other ML as RDO ones just\n \u003e by modifying BASE_URL so it\u0027d be nice to have a way to override it\n \u003e via CLI parameter.\n\nAwesome, excellent idea I\u0027ll submit a new patch set to add a new param to allow us to override the base url.","accounts_in_message":[],"_revision_number":2},{"id":"4532e994b2d06bb5934daf50239e442dd831ad6b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 15:39:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7fefd76d262846f31209f3c72ab53df5f0bc583e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 15:45:22.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n\u003e Nice tool!, i\u0027ve successfully tested with other ML as RDO ones just\n \u003e by modifying BASE_URL so it\u0027d be nice to have a way to override it\n \u003e via CLI parameter.\n\nI added the possibility to override the default url.\n\nI introduced the `--mailing-list` param which accept a valid url.\n\nExample of usage:\n\n```\n$ tools/search_emails.py --topic \".?openstack/watcher.*\" --mailing-list http://lists.openstack.org/pi\npermail/release-job-failures/ --starting-date 2016-6-1                                                                                                                                                               \n                                                                                                                                                                                                                     \nLooking for emails between 2016 June and 2020 October and sent by                                                                                                                                                    \nAnalyzing June 2016: http://lists.openstack.org/pipermail/release-job-failures//2016-June                                                                                                                            \nAnalyzing July 2016: http://lists.openstack.org/pipermail/release-job-failures//2016-July                                                                                                                            \nAnalyzing August 2016: http://lists.openstack.org/pipermail/release-job-failures//2016-August                                                                                                                        \n...\nAnalyzing September 2020: http://lists.openstack.org/pipermail/release-job-failures//2020-September\nAnalyzing October 2020: http://lists.openstack.org/pipermail/release-job-failures//2020-October\nError with http://lists.openstack.org/pipermail/release-job-failures//2020-October\n5 result(s) have been found\n2016-September:\n        - [Release-job-failures] Release of openstack/watcher failed - jenkins at openstack.org\n        http://lists.openstack.org/pipermail/release-job-failures//2016-September/000130.html\n        - [Release-job-failures] Release of openstack/watcher failed - jenkins at openstack.org\n        http://lists.openstack.org/pipermail/release-job-failures//2016-September/000176.html\n        - [Release-job-failures] Release of openstack/watcher-dashboard failed - jenkins at openstack.org\n        http://lists.openstack.org/pipermail/release-job-failures//2016-September/000175.html\n2020-April:\n        - [Release-job-failures] Release of openstack/watcher-tempest-plugin    for ref refs/tags/2.0.0 failed - zuul at openstack.org\n        http://lists.openstack.org/pipermail/release-job-failures//2020-April/001361.html\n2020-September:\n        - [Release-job-failures] Release of openstack/watcher-tempest-plugin    for ref refs/tags/2.1.0 failed - zuul at openstack.org\n        http://lists.openstack.org/pipermail/release-job-failures//2020-September/001466.html\n```\n\nNotice that it fail on October 2020 since no mail have been sent on this email this month, so the October\u0027s page doesn\u0027t exist yet and we just display a warning to inform users.\n\nDo not hesitate to tell me what do you think about this PS.","accounts_in_message":[],"_revision_number":2},{"id":"d8ed24b1e37438eb23d464681e02221146770c14","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 15:50:06.000000000","message":"Patch Set 2:\n\n\u003e Notice that it fail on October 2020 since no mail have been sent on\n \u003e this email this month, so the October\u0027s page doesn\u0027t exist yet and\n \u003e we just display a warning to inform users.\n\ns/no mail have been sent on this email this month/no mail have been sent on this ML this month/","accounts_in_message":[],"_revision_number":2},{"id":"585812315fb766feb5293802e25e693788f829a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 15:55:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae3f6c9f1ad48dba28981ecff4bad82 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/541a2ef7abab41a5b3f0ceb4a8f213af : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1defda1ef4b84efaa5b2ce15017eb27b : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe166b22b74c4f15aba7d3f13111fd5d : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3dd518ff82bf45718728cf5bac449ca7 : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f31f0acb5c424c5a8967c44146a97407 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d54ef7da25fe446f9403453c94cad23b : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"5a830d978c6c0e446f0373c7975f85811b048b7c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 09:10:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5af4cbff72c270021a480a7657cd7587876f11dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:22:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d0fb64a8cf40e3a393dffc293ff714 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7fe4f423f2d467092252fb801e5673b : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30b0f67eb651405d8d8c858f848cd0ef : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47df2492f9244a788542d334400b274 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17f3a9603c0b497e88aea16df0144599 : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f96c5aeea104416baa07c9f88611ad01 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/649b4bc44f084d16a9993677aa8ef6ed : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4},{"id":"bb1e46df956018a126b0fb19c8a66f5804ddf836","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-02 09:55:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2b50779af0396cf722f7409cb9de49ba290eb071","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-10-02 10:21:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f850e3d99df34b6924562c9b988b5d42c26163d3","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-02 11:25:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4a3dfec4056e5220bac0728416f4c71e31cd0bd7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 13:57:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bc23ab6f34c231b7da64950f64c91c01619c10f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 13:59:38.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThanks Daniel for your comments you are right it could be worth to return an exit code.","accounts_in_message":[],"_revision_number":5},{"id":"36e720534f0dca63e507492fe842896369c232b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:11:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60994575be3842648c33ab92e73ede9c : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6bbe3d758f457087a0b41eabfd5a1a : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/136a5fc5b5bf4288a513370929705131 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7affab258d824021a54bea9eade5ee3e : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bd464986ed04bf88f234c35ad4a45b0 : SUCCESS in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3076a4e6d434429eb7f7b1e909a4688c : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2acbf452e0b049c1bdec0390b22a0b68 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e80304c2a25db771be72f897e713ebb8e5441556","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-10-05 08:42:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0432ed58ae8016fb410b80c89c4ca7877851bb0f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 08:44:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"796601d1bb8b389db21d1917e5f5f8624dfc1fe3","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-05 08:51:44.000000000","message":"Patch Set 5: Code-Review+1\n\nSorry my bad. Actually, the exit code is manage in the search function.","accounts_in_message":[],"_revision_number":5},{"id":"bb4b0e1adb48fe6208820dca806deb8466c63c73","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-07 15:01:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7f25beecd28d69258915fd8e8f42f601b39689d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 16:08:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a2a3e9a806afbdcd9ff6d5122a28c9f5c075bdf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:08:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cc1b1d6a4368f370519b65707160679b2174e724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:25:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9af70b07166433ba2756c8e7ec127bb : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cfc83e7a0b14ba387d90b9a0504785f : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cd8cc22c15e4b3397b8d773ea9aa760 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab0e3ed82344a22baf4eb2d929b6df0 : SUCCESS in 9m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65d6333d7f604feb96583d63924137ed : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3ad10319bef4f0d90c4a0a7b5a32fff : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":5},{"id":"6505c158967234b524326f193c8f35e44c37f7ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:25:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8ab91eebc7a0b9d0be863772ea286627e6b57d9","revisions":{"2a78fb74a4eb33695f7bf3108b9eb1210ea33263":{"kind":"REWORK","_number":1,"created":"2020-09-30 17:59:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/62/755362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/755362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/755362/1"}}},"commit":{"parents":[{"commit":"12d50f20d696b20dd68d94ef74864a3e91d8c184","subject":"Merge \"Release trove-tempest-plugin for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12d50f20d696b20dd68d94ef74864a3e91d8c184"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 16:43:58.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 17:57:47.000000000","tz":120},"subject":"Adding a new tools to search topic on the ML","message":"Adding a new tools to search topic on the ML\n\nBy default it will research for topics related to the release team but\npattern can be overriden to looking for all kind of topics (c.f the\nrelated doc).\n\nChange-Id: Ib1fe5f74b408ac489b12c4ef003860ce7aee1532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a78fb74a4eb33695f7bf3108b9eb1210ea33263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a78fb74a4eb33695f7bf3108b9eb1210ea33263"}]},"branch":"refs/heads/master"},"9f698ef44cb86da6243affc73251a270dbd6fab3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-30 18:01:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/62/755362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/755362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/755362/2"}}},"commit":{"parents":[{"commit":"12d50f20d696b20dd68d94ef74864a3e91d8c184","subject":"Merge \"Release trove-tempest-plugin for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12d50f20d696b20dd68d94ef74864a3e91d8c184"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 16:43:58.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 18:00:08.000000000","tz":120},"subject":"Adding a new tools to search topic on the ML","message":"Adding a new tools to search topic on the ML\n\nBy default it will research for topics related to the release team but\npattern can be overriden to looking for all kind of topics (c.f the\nrelated doc).\n\nExample of usage and produced output:\n\n```\ntools/search_emails.py --topic \".?\\[release\\] Release countdown.*\" --starting-date 2020-5-1\nLooking for emails between 2020 May and 2020 September and sent by\nAnalyzing May 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-May\nAnalyzing June 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-June\nAnalyzing July 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-July\nAnalyzing August 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-August\nAnalyzing September 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-September\n15 result(s) have been found\n2020-May:\n        - [release] Release countdown for week R-1, May 4 - 8 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014577.html\n        - [release] Release countdown for week R-0, May 11 - 15 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014780.html\n        - [release] Release countdown for week R-20, May 25 - 29 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014999.html\n2020-June:\n        - [release] Release countdown for week R-17, Jun 15 - Jun 19 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015415.html\n        - [release] Release countdown for week R-16, Jun 22 - Jun 26 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015541.html\n2020-July:\n        - [release] Release countdown for week R-13  July 13 - July 17 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015878.html\n        - [release] Release countdown for week R-12  July 20 - July 24 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016056.html\n2020-August:\n        - [release] Release countdown for week R-10  August 3 - 7 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016243.html\n        - [release] Release countdown for week R-8 Aug 17 - 21 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016534.html\n        - [release] Release countdown for week R-7 Aug 24 - 28 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016709.html\n        - [release] Release countdown for week R-6 Aug 31 - Sept 4 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016894.html\n2020-September:\n        - [release] Release countdown for week R-5 Sept 7 - 11 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017079.html\n        - [release] Release countdown for week R-4 Sept 14 - 18 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017210.html\n        - [release] Release countdown for week R-3 Sept 21 - 25 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017399.html\n        - [release] Release countdown for week R-2 Sept 28 - Oct 2 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017616.html\n```\n\nChange-Id: Ib1fe5f74b408ac489b12c4ef003860ce7aee1532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f698ef44cb86da6243affc73251a270dbd6fab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f698ef44cb86da6243affc73251a270dbd6fab3"}]},"branch":"refs/heads/master"},"174c70c6bf15c5d7cd39e3cd6407209d724faa5a":{"kind":"REWORK","_number":3,"created":"2020-10-01 15:39:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/62/755362/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/755362/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/755362/3"}}},"commit":{"parents":[{"commit":"12d50f20d696b20dd68d94ef74864a3e91d8c184","subject":"Merge \"Release trove-tempest-plugin for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12d50f20d696b20dd68d94ef74864a3e91d8c184"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 16:43:58.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 15:38:56.000000000","tz":120},"subject":"Adding a new tools to search topic on the ML","message":"Adding a new tools to search topic on the ML\n\nBy default it will research for topics related to the release team but\npattern can be overriden to looking for all kind of topics (c.f the\nrelated doc).\n\nExample of usage and produced output:\n\n```\ntools/search_emails.py --topic \".?\\[release\\] Release countdown.*\" --starting-date 2020-5-1\nLooking for emails between 2020 May and 2020 September and sent by\nAnalyzing May 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-May\nAnalyzing June 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-June\nAnalyzing July 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-July\nAnalyzing August 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-August\nAnalyzing September 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-September\n15 result(s) have been found\n2020-May:\n        - [release] Release countdown for week R-1, May 4 - 8 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014577.html\n        - [release] Release countdown for week R-0, May 11 - 15 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014780.html\n        - [release] Release countdown for week R-20, May 25 - 29 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014999.html\n2020-June:\n        - [release] Release countdown for week R-17, Jun 15 - Jun 19 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015415.html\n        - [release] Release countdown for week R-16, Jun 22 - Jun 26 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015541.html\n2020-July:\n        - [release] Release countdown for week R-13  July 13 - July 17 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015878.html\n        - [release] Release countdown for week R-12  July 20 - July 24 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016056.html\n2020-August:\n        - [release] Release countdown for week R-10  August 3 - 7 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016243.html\n        - [release] Release countdown for week R-8 Aug 17 - 21 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016534.html\n        - [release] Release countdown for week R-7 Aug 24 - 28 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016709.html\n        - [release] Release countdown for week R-6 Aug 31 - Sept 4 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016894.html\n2020-September:\n        - [release] Release countdown for week R-5 Sept 7 - 11 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017079.html\n        - [release] Release countdown for week R-4 Sept 14 - 18 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017210.html\n        - [release] Release countdown for week R-3 Sept 21 - 25 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017399.html\n        - [release] Release countdown for week R-2 Sept 28 - Oct 2 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017616.html\n```\n\nChange-Id: Ib1fe5f74b408ac489b12c4ef003860ce7aee1532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/174c70c6bf15c5d7cd39e3cd6407209d724faa5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/174c70c6bf15c5d7cd39e3cd6407209d724faa5a"}]},"branch":"refs/heads/master"},"d48f86f1e1a4e863500e2c395a12c1981197884c":{"kind":"REWORK","_number":4,"created":"2020-10-02 09:10:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/62/755362/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/755362/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/755362/4"}}},"commit":{"parents":[{"commit":"12d50f20d696b20dd68d94ef74864a3e91d8c184","subject":"Merge \"Release trove-tempest-plugin for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12d50f20d696b20dd68d94ef74864a3e91d8c184"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 16:43:58.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-02 09:07:08.000000000","tz":120},"subject":"Adding a new tools to search topic on the ML","message":"Adding a new tools to search topic on the ML\n\nBy default it will research for topics related to the release team but\npattern can be overriden to looking for all kind of topics (c.f the\nrelated doc).\n\nUsages\n------\n\nThe following example will search for all emails sent openstack-discuss\nwho match release topics from the creation of the mailing list to the\ncurrent day:\n\n```\n$ tools/search_emails.py\n```\n\nIn the following example we search for release countdown emails sent\nbetween May 2020 and today:\n\n```\ntools/search_emails.py --topic \".?\\[release\\] Release countdown.*\" --starting-date 2020-5-1\nLooking for emails between 2020 May and 2020 September and sent by\nAnalyzing May 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-May\nAnalyzing June 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-June\nAnalyzing July 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-July\nAnalyzing August 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-August\nAnalyzing September 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-September\n15 result(s) have been found\n2020-May:\n        - [release] Release countdown for week R-1, May 4 - 8 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014577.html\n        - [release] Release countdown for week R-0, May 11 - 15 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014780.html\n        - [release] Release countdown for week R-20, May 25 - 29 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014999.html\n2020-June:\n        - [release] Release countdown for week R-17, Jun 15 - Jun 19 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015415.html\n        - [release] Release countdown for week R-16, Jun 22 - Jun 26 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015541.html\n2020-July:\n        - [release] Release countdown for week R-13  July 13 - July 17 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015878.html\n        - [release] Release countdown for week R-12  July 20 - July 24 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016056.html\n2020-August:\n        - [release] Release countdown for week R-10  August 3 - 7 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016243.html\n        - [release] Release countdown for week R-8 Aug 17 - 21 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016534.html\n        - [release] Release countdown for week R-7 Aug 24 - 28 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016709.html\n        - [release] Release countdown for week R-6 Aug 31 - Sept 4 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016894.html\n2020-September:\n        - [release] Release countdown for week R-5 Sept 7 - 11 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017079.html\n        - [release] Release countdown for week R-4 Sept 14 - 18 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017210.html\n        - [release] Release countdown for week R-3 Sept 21 - 25 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017399.html\n        - [release] Release countdown for week R-2 Sept 28 - Oct 2 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017616.html\n```\n\nChange-Id: Ib1fe5f74b408ac489b12c4ef003860ce7aee1532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d48f86f1e1a4e863500e2c395a12c1981197884c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d48f86f1e1a4e863500e2c395a12c1981197884c"}]},"branch":"refs/heads/master"},"a8ab91eebc7a0b9d0be863772ea286627e6b57d9":{"kind":"REWORK","_number":5,"created":"2020-10-02 13:57:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/62/755362/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/755362/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/755362/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/755362/5"}}},"commit":{"parents":[{"commit":"12d50f20d696b20dd68d94ef74864a3e91d8c184","subject":"Merge \"Release trove-tempest-plugin for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12d50f20d696b20dd68d94ef74864a3e91d8c184"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-30 16:43:58.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-02 13:57:38.000000000","tz":120},"subject":"Adding a new tools to search topic on the ML","message":"Adding a new tools to search topic on the ML\n\nBy default it will research for topics related to the release team but\npattern can be overriden to looking for all kind of topics (c.f the\nrelated doc).\n\nUsages\n------\n\nThe following example will search for all emails sent openstack-discuss\nwho match release topics from the creation of the mailing list to the\ncurrent day:\n\n```\n$ tools/search_emails.py\n```\n\nIn the following example we search for release countdown emails sent\nbetween May 2020 and today:\n\n```\ntools/search_emails.py --topic \".?\\[release\\] Release countdown.*\" --starting-date 2020-5-1\nLooking for emails between 2020 May and 2020 September and sent by\nAnalyzing May 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-May\nAnalyzing June 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-June\nAnalyzing July 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-July\nAnalyzing August 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-August\nAnalyzing September 2020: http://lists.openstack.org/pipermail/openstack-discuss/2020-September\n15 result(s) have been found\n2020-May:\n        - [release] Release countdown for week R-1, May 4 - 8 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014577.html\n        - [release] Release countdown for week R-0, May 11 - 15 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014780.html\n        - [release] Release countdown for week R-20, May 25 - 29 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014999.html\n2020-June:\n        - [release] Release countdown for week R-17, Jun 15 - Jun 19 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015415.html\n        - [release] Release countdown for week R-16, Jun 22 - Jun 26 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015541.html\n2020-July:\n        - [release] Release countdown for week R-13  July 13 - July 17 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015878.html\n        - [release] Release countdown for week R-12  July 20 - July 24 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016056.html\n2020-August:\n        - [release] Release countdown for week R-10  August 3 - 7 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016243.html\n        - [release] Release countdown for week R-8 Aug 17 - 21 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016534.html\n        - [release] Release countdown for week R-7 Aug 24 - 28 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016709.html\n        - [release] Release countdown for week R-6 Aug 31 - Sept 4 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016894.html\n2020-September:\n        - [release] Release countdown for week R-5 Sept 7 - 11 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017079.html\n        - [release] Release countdown for week R-4 Sept 14 - 18 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017210.html\n        - [release] Release countdown for week R-3 Sept 21 - 25 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017399.html\n        - [release] Release countdown for week R-2 Sept 28 - Oct 2 - Sean McGinnis\n        http://lists.openstack.org/pipermail/openstack-discuss/2020-September/017616.html\n```\n\nChange-Id: Ib1fe5f74b408ac489b12c4ef003860ce7aee1532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8ab91eebc7a0b9d0be863772ea286627e6b57d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8ab91eebc7a0b9d0be863772ea286627e6b57d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
