)]}'
{"id":"openstack%2Freleases~755522","triplet_id":"openstack%2Freleases~master~I02dd9dfccc58945348de7e4a738a9b07bae9b18a","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I02dd9dfccc58945348de7e4a738a9b07bae9b18a","subject":"New minor release for renderspec 2.1.0.","status":"MERGED","created":"2020-10-01 14:13:38.000000000","updated":"2020-10-02 12:05:35.000000000","submitted":"2020-10-02 12:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755522-1601640335009-7a2353a8","meta_rev_id":"e28cc77eb2f8e033084227a9c4c2c75433367a01","_number":755522,"virtual_id_number":755522,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 12:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-02 11:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-02 11:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-02 11:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"date":"2020-10-02 11:36:13.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 14:20:44.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-10-01 14:25:57.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-10-01 16:44:10.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-02 11:36:13.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-02 11:37:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-02 11:51:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-02 12:05:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14de7e5349c09deac1c54ed663dd36cbe4a804c7","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-10-01 14:13:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a0e304bfb83adb4bfea155bda1e540f10c7b601","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-01 14:18:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f85e8fc30a794535e5f0defe28e34965b532cc74","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-10-01 14:20:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ea9d39ca7ca2c57e942335d3b3b24f3beba0167","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-10-01 14:25:57.000000000","message":"Patch Set 1: Code-Review+1\n\nVersion is okay due to drop of 2.7.","accounts_in_message":[],"_revision_number":1},{"id":"13e08ffbb3e931b74433f770c8bd7e480ece5b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 14:27:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e3f6dfd86145f68184eebf0f5f12cc : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94953a485e7649f4a01153f9dfe469d9 : SUCCESS in 6m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/73783e2a62a649cfb71da2167e72bccd : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1564d1d8f8ad1828498dcf123492861091349bbd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 15:19:12.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"f2c79afc2d4bf0f0ee11ac50b55fda5d2f70f2e7","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-10-01 16:16:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ec75a21663781d6abaa8a8cf75b5c5d880e334e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 16:19:42.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"ecf8294e2753793818b7694ef083e7724613a320","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 16:20:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de39a34aea0598a664a4ba0e7293de8bb29c0c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:31:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8565c85bbc4047a60643447a6e5c6e : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10e7827cad8f45188df4ac9b32638ad3 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46b40bc0643241acbf0dec8788ab85f6 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"cce6ea805b1d2c4df549d87ee406e03eca843560","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-01 16:44:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc257cc1b71f16f78325502749046665f4c3a16f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 18:53:19.000000000","message":"Patch Set 2: Code-Review-1\n\nSince this library has passed the 1.x threshold - the removal of py27 support is a backwards incompatible change. Because of that, this would require a major version bump to reflect that. So this should be a 3.0.0 release.","accounts_in_message":[],"_revision_number":2},{"id":"d002d55fc45ee66977f8f6301b64616aae3bda81","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 08:13:41.000000000","message":"Patch Set 2:\n\n\u003e Since this library has passed the 1.x threshold - the removal of\n \u003e py27 support is a backwards incompatible change. Because of that,\n \u003e this would require a major version bump to reflect that. So this\n \u003e should be a 3.0.0 release.\n\nPython 2.7 is support have been already dropped and landed with 2.0.0.\n\nHere minor changes and small clean ups are embed so a minor version is really enough.\n\nThe commit message is outdated, I discussed directly with Thomas on IRC this is why he submitted the PS2 but he forgotten to update the commit_msg.","accounts_in_message":[],"_revision_number":2},{"id":"a4dfdeaba642d2797ebd67ca9ed9acce0191ced5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 08:16:55.000000000","message":"Patch Set 2: Code-Review-1\n\n-1 just to highlight the outdated commit_msg and to drop the py2.7 line.\n\n@Sean here is the list of changes who will be landed:\n\n* eb69d47 Improve distro detection for recent Fedora releases\n* 8e84ccd Switch to newer openstackdocstheme version\n* ece74aa Cleanup py27 support\n* 1910877 Python 3.8 tests are passing, declare as supported","accounts_in_message":[],"_revision_number":2},{"id":"52308938ff1a2ec1f7a68552d204c5b388b653f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-02 11:05:25.000000000","message":"Patch Set 2:\n\nLooking at the list-changes output now, I see that it really is just a commit message issue. Not as major of a problem then, but I do think it would be good to update the commit message to avoid any (further) potential confusion.","accounts_in_message":[],"_revision_number":2},{"id":"7978b7cf4d75e920cd4515169b3bbc3d112d6ca2","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-02 11:36:13.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b564f2417fefb58b8a0058f74c5ce39af3eace75","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 11:37:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2b09b8a792bc11b30c4ec356843c2cf58a96a414","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 11:37:35.000000000","message":"Patch Set 3:\n\nThanks Daniel","accounts_in_message":[],"_revision_number":3},{"id":"0169abc3213c8329c364a16eec056a9092a4cdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:48:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8aa9de14fd43e5bbfe5384d06fc8e2 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce556235e6984be897e53e0a4330036c : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/75b4cbee01b04045ad4a0c1af0267f5f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ca06143bdb55be2de162b33b344aba0841c074db","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-02 11:51:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cde2ec5cfe130c8c98db44a786153556d1d1e1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:51:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"576f03e7e7011d5b1ce3f71174dd3520838bd1fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:05:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d407c89b09404a6eb3ef0aa321ef3942 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d976baaadf454b12ba56dd16eb706067 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0803547fd1945b08b3430f9921fa707 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e28cc77eb2f8e033084227a9c4c2c75433367a01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:05:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9dd24992ed720eb8172b398e1911d3627b18860","revisions":{"75bae3fed15bcfab97bdbac5d4fd627c0d911d37":{"kind":"REWORK","_number":1,"created":"2020-10-01 14:13:38.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/22/755522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/755522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/755522/1"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-10-01 14:13:12.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-10-01 14:13:12.000000000","tz":120},"subject":"renderspec: Release 3.0.0","message":"renderspec: Release 3.0.0\n\n- Drop python 2.7 support\n- Other smaller fixes\n\nChange-Id: I02dd9dfccc58945348de7e4a738a9b07bae9b18a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75bae3fed15bcfab97bdbac5d4fd627c0d911d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75bae3fed15bcfab97bdbac5d4fd627c0d911d37"}]},"branch":"refs/heads/master"},"33b8246754e38f3e823e157105174001a4bd4213":{"kind":"REWORK","_number":2,"created":"2020-10-01 16:16:02.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/22/755522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/755522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/755522/2"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-10-01 14:13:12.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-10-01 16:15:50.000000000","tz":120},"subject":"renderspec: Release 2.1.0","message":"renderspec: Release 2.1.0\n\n- Drop python 2.7 support\n- Other smaller fixes\n\nChange-Id: I02dd9dfccc58945348de7e4a738a9b07bae9b18a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33b8246754e38f3e823e157105174001a4bd4213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33b8246754e38f3e823e157105174001a4bd4213"}]},"branch":"refs/heads/master"},"f9dd24992ed720eb8172b398e1911d3627b18860":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-02 11:36:13.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/22/755522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/755522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/755522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/755522/3"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2020-10-01 14:13:12.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-02 11:40:41.000000000","tz":120},"subject":"New minor release for renderspec 2.1.0.","message":"New minor release for renderspec 2.1.0.\n\nSmaller fixes.\n\nChange-Id: I02dd9dfccc58945348de7e4a738a9b07bae9b18a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9dd24992ed720eb8172b398e1911d3627b18860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9dd24992ed720eb8172b398e1911d3627b18860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
