)]}'
{"id":"openstack%2Freleases~755593","triplet_id":"openstack%2Freleases~master~I7c6bd312c58c1b701936403c8500582bc177b3fb","project":"openstack/releases","branch":"master","topic":"victoria-branch","hashtags":[],"change_id":"I7c6bd312c58c1b701936403c8500582bc177b3fb","subject":"Branch rpm-packaging for Victoria","status":"MERGED","created":"2020-10-01 16:49:34.000000000","updated":"2020-10-06 14:02:25.000000000","submitted":"2020-10-06 14:02:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755593-1601992945677-ca15d9bd","meta_rev_id":"edff3f36ef96b109dcdbe52a1ff5700142e11b3d","_number":755593,"virtual_id_number":755593,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 14:02:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-06 13:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-06 08:56:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-05 17:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-06 08:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-06 13:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 08:09:15.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-06 08:20:39.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-10-06 08:56:49.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-10-06 13:48:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-06 14:02:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3efa6cf0c05338b47740b312a827c4ae1dfe62d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 16:49:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1937f9a02825221804a42bc06f8d63e4f976389d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 16:50:03.000000000","message":"Patch Set 1: Workflow-1\n\nPlease hold it until monday","accounts_in_message":[],"_revision_number":1},{"id":"940aabccbb2d2a2773ad6a484889ebad0445f426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d95758446c4adb9739e1ef3d0f080f : SUCCESS in 9m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3b0cc810b6d4559a9be2236cc2deecf : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a094fedbfdbb440c862c0e397f32a6e8 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7a11acc0c6e5d9530fb83bd42b839f64fa312bb2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 18:46:33.000000000","message":"Patch Set 1:\n\nThis deliverable file should actually be removed. rpm-packaging is now a SIG, so the release team no longer manages releases for it.","accounts_in_message":[],"_revision_number":1},{"id":"93ff5643735adeb2ecf015697880afaa365a530f","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-02 10:13:43.000000000","message":"Patch Set 1:\n\nIs the move to SIG already effective ? I understood that is was for W release. Here is the latest branch managed by release team.","accounts_in_message":[],"_revision_number":1},{"id":"f7b0d1ff3a7af1e6da98ad2cb114004adbcfed63","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-10-03 08:15:40.000000000","message":"Patch Set 1:\n\nI\u0027d appreciate if we could treat this as the final exception, as it will likely take days to get the permissions set up for us to be able to branch directly.","accounts_in_message":[],"_revision_number":1},{"id":"4740399e3ddb14947d83df8e82ff4c7d09655ca1","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-10-03 08:16:14.000000000","message":"Patch Set 1:\n\nalso, fwiw, I agree with Boyron that we believed the conversion would become effective for the W cycle.","accounts_in_message":[],"_revision_number":1},{"id":"8f82190f3070db5a27c440869ba4174b91d5d737","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-03 11:40:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d79f3a055dd403119123dbbdcf557d91f0db320","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 06:06:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b7a173276c61ae25580d3f070698e0d988f4113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 06:26:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f491dacf654dc89c2b144d14bf7bc3 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2432b2fe5cb14c38a97f6418c75bfea7 : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0241b92ba3b445f69bea5da90108b085 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b7de6981ed764a48ffe6ed4a8f49af05c2b1cc2f","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 07:51:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"478f1aa6b2bc592b979891d8e1677a8a3d737d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:05:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b20e2412f34431a0d3607fd3e53370 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a029aa2141c44e6a091ec576d2e490f : SUCCESS in 7m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fce4a6d98d984638832d5dc01be3b3b1 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c0d9af8cac73c3ae49e9f682a58ee4378815f4f3","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 08:05:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11e4427f795bed5d6652fe783edfe59113dc9de9","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 08:09:15.000000000","message":"Patch Set 4: Code-Review-1\n\nWe are waiting few additional hours to include latest patch before creating the branch.","accounts_in_message":[],"_revision_number":4},{"id":"51b414e70594d43a307cf23233a9aacb1cfe6291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:17:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab44da56eba498bbff8b5e055c1edbc : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68b43079a4e744d2a84fcaf07c9ffb36 : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8e3dc6c42a654422bf2b56f850c21398 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"e4aeb1737182b7cb09f9322efcb7522ee528a7ea","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 13:35:00.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9cee29ef9061bb59d5ffa42fba5c2b21c24b06b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 13:58:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec9c9548f1c45439ba66024485e3f59 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/743f8aafc5664bac8ceb66c93f81cf90 : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6c5d1ce3449e400e859ce85ce5053a53 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"1cec9b9b2d9a769fd347b4c11a3f035446a18a40","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-05 16:20:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e62ee5ee827ec8d1b192154b54b1372fa2b0e8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:45:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0267e22df744076adb755d359ce858a : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9c189f77eee4c0ebf290114d65aa366 : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7ebbeadb807a4a24a3b39f6b56da9dea : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":6},{"id":"6304fdb09b462169825e5393c7ba46ed755f1877","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 17:13:42.000000000","message":"Patch Set 6: Code-Review+2\n\nI think we can go ahead now with that, some patches [1][2] are still for waiting for merge but we need to fix some issues first, we could backport these changes in few days.\n\n[1] https://review.opendev.org/#/c/755159/\n[2] add new packages for validation-libs and validation-commons","accounts_in_message":[],"_revision_number":6},{"id":"932d42b72155944fcdb0b7a66c4c8011bda30561","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-10-06 08:20:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"dbfd0d197dc809f53c07cda6eec493940a0e886f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-10-06 08:56:49.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"cba3a9c8656a808246693e07058a496f329fb1b8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-06 13:48:04.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7a25f0c5c302976c0783beb8930b309ae1f9fc64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 13:52:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ae22f6e020f518a257450edd0735c66c4a19c801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:02:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981fc6dda4504270a03115f939b5c4a6 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5af1e60d19542708331b8a367008080 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9eb61d080857411ba6bcae8a8e48edaa : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":6},{"id":"77933c9dd32a013f8af674529d68a11a344e7aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:02:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"57a53376c9a4131ad862ed401bf2ee5e90aa87a9","revisions":{"263dda3de74352a8a00f37bcbb7c565b201150de":{"kind":"REWORK","_number":1,"created":"2020-10-01 16:49:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/755593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/1"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/263dda3de74352a8a00f37bcbb7c565b201150de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/263dda3de74352a8a00f37bcbb7c565b201150de"}]},"branch":"refs/heads/master"},"42fba8603755045a23c4c3321041ef37b84d73a8":{"kind":"REWORK","_number":2,"created":"2020-10-05 06:06:16.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/93/755593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/2"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-10-05 06:06:12.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42fba8603755045a23c4c3321041ef37b84d73a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42fba8603755045a23c4c3321041ef37b84d73a8"}]},"branch":"refs/heads/master"},"59b1d7f4fa0c195667e43124687abf83c06cf826":{"kind":"REWORK","_number":3,"created":"2020-10-05 07:51:51.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/93/755593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/3"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-10-05 07:51:23.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59b1d7f4fa0c195667e43124687abf83c06cf826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59b1d7f4fa0c195667e43124687abf83c06cf826"}]},"branch":"refs/heads/master"},"2aae1f3584dcac8332fca5b8aaf5c5d15e4b01b7":{"kind":"REWORK","_number":4,"created":"2020-10-05 08:05:40.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/93/755593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/4"}}},"commit":{"parents":[{"commit":"bc4899819f90197494ad81d20a7e62188c8e024a","subject":"Create stable/victoria for grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc4899819f90197494ad81d20a7e62188c8e024a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-10-05 08:05:21.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aae1f3584dcac8332fca5b8aaf5c5d15e4b01b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aae1f3584dcac8332fca5b8aaf5c5d15e4b01b7"}]},"branch":"refs/heads/master"},"cc5a21f7f3db3be6b66740028a96a65b6a5f1dd2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-05 13:35:00.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/93/755593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/5"}}},"commit":{"parents":[{"commit":"5392d1725a74d45c697c9193d110e8ce14704dda","subject":"fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5392d1725a74d45c697c9193d110e8ce14704dda"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-10-05 13:34:40.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc5a21f7f3db3be6b66740028a96a65b6a5f1dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc5a21f7f3db3be6b66740028a96a65b6a5f1dd2"}]},"branch":"refs/heads/master"},"57a53376c9a4131ad862ed401bf2ee5e90aa87a9":{"kind":"REWORK","_number":6,"created":"2020-10-05 16:20:43.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/93/755593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/755593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/755593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/755593/6"}}},"commit":{"parents":[{"commit":"5392d1725a74d45c697c9193d110e8ce14704dda","subject":"fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5392d1725a74d45c697c9193d110e8ce14704dda"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-01 16:49:07.000000000","tz":120},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-10-05 16:20:04.000000000","tz":120},"subject":"Branch rpm-packaging for Victoria","message":"Branch rpm-packaging for Victoria\n\nChange-Id: I7c6bd312c58c1b701936403c8500582bc177b3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/57a53376c9a4131ad862ed401bf2ee5e90aa87a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/57a53376c9a4131ad862ed401bf2ee5e90aa87a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
