)]}'
{"id":"openstack%2Freleases~755665","triplet_id":"openstack%2Freleases~master~I58f6e3de5207be8731b750e8f74f75fba3d2526e","project":"openstack/releases","branch":"master","topic":"cycle-highlights","hashtags":[],"change_id":"I58f6e3de5207be8731b750e8f74f75fba3d2526e","subject":"Add examples of previous highlights","status":"MERGED","created":"2020-10-01 22:32:20.000000000","updated":"2020-10-19 13:13:54.000000000","submitted":"2020-10-19 13:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755665-1603113234676-eaa21e4c","meta_rev_id":"a3ec1828ec4b943da10a8421a6e35be9db5b9944","_number":755665,"virtual_id_number":755665,"owner":{"_account_id":32480,"name":"Prithvi Patel","email":"pri281097@gmail.com","username":"prithvipatl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 13:13:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2020-10-05 08:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-10-07 15:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-19 13:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-10-19 13:00:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 22:43:27.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-10-05 08:46:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-10-07 15:00:50.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-19 13:00:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-19 13:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7d1eb8f403815380e22f7ba5c8338df6f076127","author":{"_account_id":32480,"name":"Prithvi Patel","email":"pri281097@gmail.com","username":"prithvipatl"},"date":"2020-10-01 22:32:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bcfb9562bd6472308562a6835748fcd72842f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:39:52.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/069bb8cc5aa248b899bd3cf362f9dbe8 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4b51833cabaae74956ebd6608e94d9084f72b215","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 07:52:28.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello,\n\nThanks for these changes.\n\nPlease see my inline comment.","accounts_in_message":[],"_revision_number":1},{"id":"7efcc9d54a15c29eaa605207423726a238459b09","author":{"_account_id":32480,"name":"Prithvi Patel","email":"pri281097@gmail.com","username":"prithvipatl"},"date":"2020-10-02 18:58:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc6208cf55cb0731589edb392377a578ffe44982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 19:09:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2caa8b655848f688dd5812ae4ae66e : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f80e16eebb26857d87e0bda2a8ed4e0554909434","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-10-05 08:46:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f35d43b1acf03069a73b4a4c09b8ca49c55a095e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-07 15:00:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d029aea38e71205b290473c0eac5c488bad23a32","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-19 13:00:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"849f32db69a00b17760b9e0bee43f9d8b2e55ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:00:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"98bb436ec34a9e7dd31db7276b99dc243e1f3b7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a0f516db6f9b72f59366d078654c2f630c2b4ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:13:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc0aac128a342e3be8e9410cc1e74a0 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342e131f928643a999f45a5557a45490 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"67da9f0a354695ba615b14edc4c1078bd1b52890","revisions":{"6cdbb60ecb13921c3696f17cb81a284e1abe7f1d":{"kind":"REWORK","_number":1,"created":"2020-10-01 22:32:20.000000000","uploader":{"_account_id":32480,"name":"Prithvi Patel","email":"pri281097@gmail.com","username":"prithvipatl"},"ref":"refs/changes/65/755665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/755665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/755665/1"}}},"commit":{"parents":[{"commit":"8ad81bf6ae43635faa74f67b3259152a2f43b3d9","subject":"Merge \"Release governance 0.8.0 for October elections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ad81bf6ae43635faa74f67b3259152a2f43b3d9"}]}],"author":{"name":"Prithvi Patel","email":"pri281097@gmail.com","date":"2020-10-01 22:29:44.000000000","tz":-240},"committer":{"name":"Prithvi Patel","email":"pri281097@gmail.com","date":"2020-10-01 22:32:19.000000000","tz":-240},"subject":"Add examples of previous highlights","message":"Add examples of previous highlights\n\nAdded Train cycle highlights for future highlights.\n\nChange-Id: I58f6e3de5207be8731b750e8f74f75fba3d2526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6cdbb60ecb13921c3696f17cb81a284e1abe7f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6cdbb60ecb13921c3696f17cb81a284e1abe7f1d"}]},"branch":"refs/heads/master"},"67da9f0a354695ba615b14edc4c1078bd1b52890":{"kind":"REWORK","_number":2,"created":"2020-10-02 18:58:10.000000000","uploader":{"_account_id":32480,"name":"Prithvi Patel","email":"pri281097@gmail.com","username":"prithvipatl"},"ref":"refs/changes/65/755665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/755665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/755665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/755665/2"}}},"commit":{"parents":[{"commit":"8ad81bf6ae43635faa74f67b3259152a2f43b3d9","subject":"Merge \"Release governance 0.8.0 for October elections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ad81bf6ae43635faa74f67b3259152a2f43b3d9"}]}],"author":{"name":"Prithvi Patel","email":"pri281097@gmail.com","date":"2020-10-01 22:29:44.000000000","tz":-240},"committer":{"name":"Prithvi Patel","email":"pri281097@gmail.com","date":"2020-10-02 18:57:48.000000000","tz":-240},"subject":"Add examples of previous highlights","message":"Add examples of previous highlights\n\nAdded Train cycle highlights for future highlights.\n\nChange-Id: I58f6e3de5207be8731b750e8f74f75fba3d2526e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67da9f0a354695ba615b14edc4c1078bd1b52890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67da9f0a354695ba615b14edc4c1078bd1b52890"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
