)]}'
{"id":"openstack%2Freleases~756015","triplet_id":"openstack%2Freleases~master~I5841aacea2134850316c8cdc5b9a27bd5d114fa6","project":"openstack/releases","branch":"master","topic":"puppet-openstack-victoria","hashtags":[],"change_id":"I5841aacea2134850316c8cdc5b9a27bd5d114fa6","subject":"Create stable/victoria branches for Puppet OpenStack","status":"MERGED","created":"2020-10-05 08:54:42.000000000","updated":"2020-10-08 14:29:03.000000000","submitted":"2020-10-08 14:29:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756015-1602167343722-8f1aa1ae","meta_rev_id":"b97985e73c28ab9e80c60b857fd299a159db5c07","_number":756015,"virtual_id_number":756015,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 14:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-08 14:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-08 08:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-10-08 05:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-10-07 23:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-07 14:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-10-06 13:10:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-08 14:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-10-07 23:41:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 09:23:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-05 11:03:17.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-06 13:10:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-07 14:00:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-07 23:40:55.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-08 05:16:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-10-08 08:36:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-10-08 14:18:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-08 14:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0e08d16d734432a38135bbf15bf65896895a847","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 08:54:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ad44ddbb568ec27fc1277960bc56f3ce66ebec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 08:56:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b9be60aa519222f255183a0f61ec07fb76e7fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:11:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4a8351f6964b2d9e14314fc8efa517 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72df42c4946048148adbf09ba5029ac6 : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/949026250d8f43fb92921befbf6c3d27 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1467bdf32918326daabdccd8e41b9fcb962d0a56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 09:21:21.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, waiting for PTL\u0027s approval","accounts_in_message":[],"_revision_number":2},{"id":"69e2758d47fb7e5ef6ec8311091a05cbeb09b61e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-10-05 09:23:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1d8c3f5f4af2c9536a7d0062645af2c62eaf7064","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 10:59:12.000000000","message":"Patch Set 2: Workflow-1\n\nOne problem I found is that current RC1 hash misses several patches TripleO, which depends on Puppet OpenStack, expect to be included in Victoria.\n - puppet-nova: https://review.opendev.org/#/c/748666/ (This is also required by current openstack-integration )\n - puppet-glance: https://review.opendev.org/#/c/750950/\n\nAlso, there are some deprecation patch which would be worth getting merged in Victoria, to avoid delaying deprecation period.\n - puppet-aodh: https://review.opendev.org/#/c/755140/\n\nI\u0027ll propose x.4.1 release for these repos and then create stable/victoria based on x.4.1, but since I don\u0027t see any existing releases from x.4.1, I\u0027d like to hear some opinions from this direction.","accounts_in_message":[],"_revision_number":2},{"id":"58fb45650c85119c4a174a433d134e3d8ba9be2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 11:11:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"165d225c3d80e666a74558162a2913fe791e71cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 11:18:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"98656571f3758465f8f2cec2b1108590acf92adb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 11:19:10.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"617d4d06f80114d30b565b2565cdde4fd696d411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 11:32:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f136340318d49ceb99f5d8e1ef35a7a : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/087138e528924aeaae6342ae7e782baa : FAILURE in 6m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0a98d04f34f648b88f2d726626928386 : FAILURE in 5m 50s","accounts_in_message":[],"_revision_number":5},{"id":"2b170a3cd7108b29ec8f0e30a7d0d5051abb0288","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 16:13:24.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"11af96c238e106643da6d9ac1e3531b5b7279759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:25:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de4eaf7afab43ed82820080e3c7ea19 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e380851e66be43eca18138ca0340b0b2 : FAILURE in 8m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4819dd0df58b413bbefe6be362fdce86 : FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":6},{"id":"6f5896bdb1642cd8d89c56d25f396528651f978e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 23:12:32.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c88517d8b74b0165d1c7ac8d8ce00fcd8a7a2d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:31:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8131ff3a22454e6caba33c1e6d447ea9 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e83ac723a2d48e98bcbf9f409e69ce4 : FAILURE in 7m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5c90943126eb4217b4c23797cd175bd3 : FAILURE in 4m 39s","accounts_in_message":[],"_revision_number":7},{"id":"3b980c9200c828b3ef0eba77f07bc7e183245c83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 02:51:53.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"cb813fa46a5eafab99c7c7bbf6955783ee072bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 03:00:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/993d3045740e4cb287c0e24f85c14d6f : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/898af0bbf52f48faa4f0eacd16f3cb92 : SUCCESS in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5c1c4660149f46f9a3e6f350359b2922 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":7},{"id":"969d8f453d44abd139c7008ef85668dabbe52ec5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 10:44:49.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nHello,\n\nSee my inline comments.","accounts_in_message":[],"_revision_number":7},{"id":"818fbee05126a76ec4706a2828ae8b7f3817f7e1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 11:59:38.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"874461f88b7e2e535bf7787c25705f711393d920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:12:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c324446568a46f09a8908d98806615a : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1874884bc6994f2db56cc6d5c068cae8 : FAILURE in 7m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b00932a9ad904ba2a270209d180f84c7 : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":8},{"id":"c1dbd980788e4cb43ff85bc38a6a40c561e857d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 12:50:53.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"956e91dc2ab934f71bcabe78c74ceebddeb60845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 13:05:38.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea7e819c78d493fb6dd18019f534495 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/905b61b2e80444968cbf88971135100e : FAILURE in 8m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cfd1177cda1d4fa8ac9dc005aba7bdd8 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":8},{"id":"41d92323569e32e1248d1abcecf4b6c2cbf95ea2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 13:08:27.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"7dce400dd276dda540f9bb7df0f82b8719706b69","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 13:09:25.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ba70d9f2ebb6c1d4ada355f7f01d312d0864f397","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 13:10:11.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":9},{"id":"f61c0f8c0ff986f9cb1cb1383e7b87545c839795","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 13:10:41.000000000","message":"Patch Set 9: Code-Review+1\n\nOnly +1 by waiting PTL\u0027s approval","accounts_in_message":[],"_revision_number":9},{"id":"bb56f79ce5769edf161068e042d4c054a8a1b826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 13:20:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81c95a232864910b242c8927f983650 : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/588f59e45782499382028e4407a96b5c : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cb12da871b0044349390fa78a66f5f3b : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":9},{"id":"aca0bd291d300b1cd59bb9f5f977e7a814d8c7e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-07 14:00:00.000000000","message":"Patch Set 9: Code-Review+1\n\nThanks Takashi","accounts_in_message":[],"_revision_number":9},{"id":"64f182372c99a44a0f6cc744ccb4fc333f2a2775","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-07 23:40:55.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"2482373da75bdb8cb184b2da1313c2d6f9b181bc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 23:41:54.000000000","message":"Patch Set 9: PTL-Approved+1","accounts_in_message":[],"_revision_number":9},{"id":"67570a1227321e464bc363c99c8684884802159e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-10-08 05:16:53.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e8bcee4dbad950b1686a2a5c56fff08a003a7d93","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-10-08 08:36:50.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ce3e1a74a93f322b61968600671ac9f654b37909","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-08 14:18:03.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"5277e6d8d2fdcc2ef0c45ecb7495f10afdbd7a28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:18:15.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d025a0e5245a925147f34168328c1d505fffc2fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:29:03.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7fddcd5bd246aaa27bd768453b09f8 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3a8f3cf78746bcbbfb61d4ff209d9a : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daa99f3b27014ce3aaf082a955d5f72f : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":9},{"id":"63083e4d9b2aa8c9bde386c8bc3266894793d556","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:29:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"658bd136b190e71ec381b92735a4dde7abb8be2d","revisions":{"d152b703ee9bbca32b8656205456c743d47cc002":{"kind":"REWORK","_number":1,"created":"2020-10-05 08:54:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/1"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d152b703ee9bbca32b8656205456c743d47cc002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d152b703ee9bbca32b8656205456c743d47cc002"}]},"branch":"refs/heads/master"},"b333ea8877a426a01f89f146c86d2fba85bd00dc":{"kind":"REWORK","_number":2,"created":"2020-10-05 08:56:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/2"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:56:22.000000000","tz":540},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b333ea8877a426a01f89f146c86d2fba85bd00dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b333ea8877a426a01f89f146c86d2fba85bd00dc"}]},"branch":"refs/heads/master"},"90dbe0e0851c01a25589d95102bb483e2f2db553":{"kind":"REWORK","_number":3,"created":"2020-10-05 11:11:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/3"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 11:11:20.000000000","tz":540},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\n... and create x.4.1 release for the modules which has several fixes\nmerged since RC tag was created.\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90dbe0e0851c01a25589d95102bb483e2f2db553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90dbe0e0851c01a25589d95102bb483e2f2db553"}]},"branch":"refs/heads/master"},"fcd5db1aebcf98435b1d74c5077a482f153995b5":{"kind":"REWORK","_number":4,"created":"2020-10-05 11:18:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/4"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 11:18:33.000000000","tz":540},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates x.4.1 release for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-panko\n - puppet-nova\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fcd5db1aebcf98435b1d74c5077a482f153995b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fcd5db1aebcf98435b1d74c5077a482f153995b5"}]},"branch":"refs/heads/master"},"d780a983dcf50fb1d35513ad471ff877646f4198":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-10-05 11:19:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/5"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 11:18:59.000000000","tz":540},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates x.4.1 release for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-nova\n - puppet-panko\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d780a983dcf50fb1d35513ad471ff877646f4198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d780a983dcf50fb1d35513ad471ff877646f4198"}]},"branch":"refs/heads/master"},"3eed680209137b5d7945d72db5e851f76a5106c1":{"kind":"REWORK","_number":6,"created":"2020-10-05 16:13:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/6"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 16:13:21.000000000","tz":0},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates x.4.1 release for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-nova\n - puppet-panko\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3eed680209137b5d7945d72db5e851f76a5106c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3eed680209137b5d7945d72db5e851f76a5106c1"}]},"branch":"refs/heads/master"},"d3a485c0016316315f30f9b46449004a20009eba":{"kind":"REWORK","_number":7,"created":"2020-10-05 23:12:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/7"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 23:12:29.000000000","tz":0},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates x.4.1 release for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-nova\n - puppet-panko\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3a485c0016316315f30f9b46449004a20009eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3a485c0016316315f30f9b46449004a20009eba"}]},"branch":"refs/heads/master"},"150a4dadb4ca8e860bf7cfc09767790a5e62c3ae":{"kind":"REWORK","_number":8,"created":"2020-10-06 11:59:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/8"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-06 11:59:20.000000000","tz":0},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates new releases for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-neutron\n - puppet-nova\n - puppet-panko\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/150a4dadb4ca8e860bf7cfc09767790a5e62c3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/150a4dadb4ca8e860bf7cfc09767790a5e62c3ae"}]},"branch":"refs/heads/master"},"658bd136b190e71ec381b92735a4dde7abb8be2d":{"kind":"REWORK","_number":9,"created":"2020-10-06 13:08:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/756015/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/756015/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/756015/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/756015/9"}}},"commit":{"parents":[{"commit":"0a021211a20e4a972dcaf22979b48912ec0810a7","subject":"Merge \"New bugfix release oslo.privsep for train.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a021211a20e4a972dcaf22979b48912ec0810a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 08:54:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-06 13:08:23.000000000","tz":0},"subject":"Create stable/victoria branches for Puppet OpenStack","message":"Create stable/victoria branches for Puppet OpenStack\n\nThis change creates stable/victoria branch from RC1 release created by\nthe previous change[1]\n [1] https://review.opendev.org/#/c/754594/\n\nNote that this change creates new releases for the following modules,\nbecause these modules has several changes merged since RC tag patch was\nmerged.\n - puppet-aodh\n - puppet-cloudkitty\n - puppet-glance\n - puppet-neutron\n - puppet-nova\n - puppet-panko\n - puppet-sahara\n - puppet-swift\n\nChange-Id: I5841aacea2134850316c8cdc5b9a27bd5d114fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/658bd136b190e71ec381b92735a4dde7abb8be2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/658bd136b190e71ec381b92735a4dde7abb8be2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
