)]}'
{"id":"openstack%2Freleases~756918","triplet_id":"openstack%2Freleases~master~I1d661095847898ad0ab4d85819a6f86ca9e155ff","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I1d661095847898ad0ab4d85819a6f86ca9e155ff","subject":"Release ovsdbapp 1.6.0","status":"MERGED","created":"2020-10-08 19:00:36.000000000","updated":"2020-10-15 19:01:44.000000000","submitted":"2020-10-15 19:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756918-1602788504730-f8940f66","meta_rev_id":"69b8cf0ae88e893291852aba7a55165bb98dd0fd","_number":756918,"virtual_id_number":756918,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 19:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-10-15 15:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-10-15 18:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-15 14:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-10-15 17:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-10-15 18:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-10-15 14:52:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 19:20:08.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-10-15 14:51:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-15 15:49:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-15 17:47:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-15 18:53:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-15 19:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2df0b04e63199c470c2b194ae191cf5b19280071","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-08 19:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41e32f049bdc572fcf2cb75e5d73a7d0fbcc822a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-08 19:01:05.000000000","message":"Patch Set 1: Workflow-1\n\nThis release shouldn\u0027t happen until after the coordinated release on Wed.","accounts_in_message":[],"_revision_number":1},{"id":"c0a16c22c07dcf5eb060d6f4fd5d0095029c1f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:13:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8c614901cb49a794e54064bf85e850 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09d0dc8ba6c74a628c47fcf06a577189 : FAILURE in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/953ab002a32b41d280e5b0c7ad3173fc : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1bcd3fdf4bd0527bf568213be48152ef35a818f5","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-08 19:39:09.000000000","message":"Patch Set 1:\n\nOn the \"Changed supported versions for dependency ovs from \u003e\u003d2.8.0 to \u003e\u003d2.10.0 without at least incrementing minor number\" error, we could certainly bump again to 1.6.0 if that\u0027s the easiest way to handle it, but really this ovs version requirement goes all the way back to the 1.3.0 release. It\u0027s just that the feature we relied on in ovs for 1.3.0 was over 2 years old when we started using it in ovsdbapp and no one caught that the lower constraint was even older.","accounts_in_message":[],"_revision_number":1},{"id":"7bd858614590820f3387c6c33d7865fa2665fe7f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 21:27:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ee2858965c039a77585bfacd98d4fffab7d80e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 21:48:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31127f2270e24379a70a93fd54a0ef5a : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50551d4b0efd43f780d1324dce500378 : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/78d2535200da46d880749a76faf2a07f : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"fedf0fb01f371a1cfe47ff9d2b30b88c3723e4b3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-15 01:36:22.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"52314b6f44881f64c4441bc987c2801c5679d793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:45:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647f0616911746a48213989d75da1c5c : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b3dd888d97845e3b0eb7aed57e0b98d : SUCCESS in 4m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/78e47f862d8e4e50ba73a0a5fbe3547d : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"39cccb1d64816e62ae26d0831bdc44db90432c1e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-15 14:51:02.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d42bccf557c4b492c0c00bb8f6f273748d1c86e5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:52:46.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"06c511a865d864c8885d0f41624c58cdf8f9449f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-15 15:49:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b71de6a906a5687820c3fa2633d6afa54ac0ed7a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 17:47:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a5122c041e908d9666726985caee5ae5ec7d13d1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-15 18:53:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bfe44a148c58cdf34502689121d22a7b9148d423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:53:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bfa14ee6a9ffc1eee744525ed079d54f30fcd8de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 19:01:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"69b8cf0ae88e893291852aba7a55165bb98dd0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 19:01:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e371361c8843ad8ecfb611b82c779f : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace23148dd504636b82d20df7c0c37b7 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9554741e4a834c6198ba0069c17b0557 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8779d7453c67816e2c50313793d6a74e191a5223","revisions":{"52f6edb91a7c63c600cc42e5a548a1000047528d":{"kind":"REWORK","_number":1,"created":"2020-10-08 19:00:36.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/18/756918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/756918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/756918/1"}}},"commit":{"parents":[{"commit":"859964086d798f333798e720e260a8aa792eb0e2","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/859964086d798f333798e720e260a8aa792eb0e2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-08 18:58:18.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-08 18:58:18.000000000","tz":-300},"subject":"Release ovsdbapp 1.5.1","message":"Release ovsdbapp 1.5.1\n\nChange-Id: I1d661095847898ad0ab4d85819a6f86ca9e155ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52f6edb91a7c63c600cc42e5a548a1000047528d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52f6edb91a7c63c600cc42e5a548a1000047528d"}]},"branch":"refs/heads/master"},"0c1b3e9f0c05a568d25f5ea69e0f19b39549e652":{"kind":"REWORK","_number":2,"created":"2020-10-14 21:27:51.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/18/756918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/756918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/756918/2"}}},"commit":{"parents":[{"commit":"859964086d798f333798e720e260a8aa792eb0e2","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/859964086d798f333798e720e260a8aa792eb0e2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-08 18:58:18.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 21:26:36.000000000","tz":0},"subject":"Release ovsdbapp 1.5.1","message":"Release ovsdbapp 1.5.1\n\nChange-Id: I1d661095847898ad0ab4d85819a6f86ca9e155ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c1b3e9f0c05a568d25f5ea69e0f19b39549e652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c1b3e9f0c05a568d25f5ea69e0f19b39549e652"}]},"branch":"refs/heads/master"},"8779d7453c67816e2c50313793d6a74e191a5223":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-15 01:36:22.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/18/756918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/756918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/756918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/756918/3"}}},"commit":{"parents":[{"commit":"859964086d798f333798e720e260a8aa792eb0e2","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/859964086d798f333798e720e260a8aa792eb0e2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-08 18:58:18.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-15 01:36:17.000000000","tz":0},"subject":"Release ovsdbapp 1.6.0","message":"Release ovsdbapp 1.6.0\n\nThis is not a feature release, but it does contain a patch that,\nbelatedly, bumps the lower requirement of the python ovs package. That\nbump really should have happened in the 1.3.0 release.\n\nChange-Id: I1d661095847898ad0ab4d85819a6f86ca9e155ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8779d7453c67816e2c50313793d6a74e191a5223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8779d7453c67816e2c50313793d6a74e191a5223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
