)]}'
{"id":"openstack%2Freleases~758378","triplet_id":"openstack%2Freleases~master~I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827","project":"openstack/releases","branch":"master","topic":"release-inbox","hashtags":[],"change_id":"I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827","subject":"Managing the review inbox by series","status":"ABANDONED","created":"2020-10-15 09:39:03.000000000","updated":"2020-10-26 13:11:11.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e5b3353ce3e263b870107cfbfcc04423b6be0916","_number":758378,"virtual_id_number":758378,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-22 16:13:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-26 12:43:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 12:24:41.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-19 16:10:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-22 16:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 12:43:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"7243292d92b7fe02243daac79ac6348fc70c0b25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 09:39:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d1e25541facaf57aaff245676aa11c108ea85d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc76db512b74cabb1c14b0ec50a3f59 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"72f91c0f4f7b74c7ea260f7249b98993fa318084","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-15 12:24:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bb90e214a8e3cda115e37c38fab4eb26154bc785","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-10-15 14:06:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"beaa8459e1f3671ba6bf4a77bd51d40da9f71466","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 14:13:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82d1b629bee34757ab0e925864a1cf425623b285","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 14:14:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c638e3a5cfcd503bc94582f7cef45f7b95ab72d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:22:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bcf0359e52419d9a51cf2a6a30546a : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c21ebf3408130052a86c03b69b5a908a6a48576b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 14:56:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4bd04ba9274f1a3291450be96cba3093bfd55e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 15:11:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049a97a5f6aa4a3394c8f0b398d3b522 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5211c17df062eb630d8f19be448fde0563763f27","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-15 15:16:33.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"12110ee288d8e4e0d16fe1c65a3583cf09a7d16d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-16 09:36:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"90affd562beba333dea3242e2eb8a41a5c2e6188","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-16 09:41:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4582caec74445b76a70afe64a29e9e7642b226d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-16 09:48:04.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"57a66d021267326bdc6ae0d30d9a4cff8b76434b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:50:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b95ff2b22924c2ab454e659e5f2be21 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":6},{"id":"06d93e105266465b5052d2241220b9ad397ed392","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-16 09:52:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8184e23f67c64626d3563b507cf2e50d3a39533","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 16:10:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e681fa92548c8071ae18cc323ab3dccbe3ae923a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 14:37:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1457f7be199f27001ec1674e2f9ce882e5b92bf6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 14:45:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9e28be09ad36ed1d608443e8cb2e3c20ed1f8c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 14:45:46.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3a06081939409f95806576cfd54447 : FAILURE in 6m 33s","accounts_in_message":[],"_revision_number":7},{"id":"8473f0e36f0f721a540d702345492caf6859353c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 14:50:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b886d93dc374ad6deb5b402411387a8fbce702b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 15:03:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1aa2b3281a4eaa84c22b02fb57d13c : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":8},{"id":"e8c191111852708a5c49a9f84d60a0c3a99af46c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 15:55:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"13b91bddc1046f743712a3bc0b108ef7acdb3c38","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 15:56:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e4c151f8ed61ffe7b65b283643c7ee76e0ae2aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:13:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9102c4eb04fd42ce8b138d9c3c27d620 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":10},{"id":"d19c01ab876883e5b311a94351a1c760b6b0ace0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-10-26 12:43:57.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"43e7dd4210a870e3ea16b84d2bca1eeb46d47ac4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-26 13:11:11.000000000","message":"Abandoned\n\nSean proposed a better approach [1] to manage these needs so I abandon these changes\n\n[1] https://review.opendev.org/#/c/759309/","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"af777a9034e814622d34be96a374a5be7bb9b45f","revisions":{"6d6c0d727700990bbdf2b3f026b4581cfb71e426":{"kind":"REWORK","_number":1,"created":"2020-10-15 09:39:03.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/1"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d6c0d727700990bbdf2b3f026b4581cfb71e426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d6c0d727700990bbdf2b3f026b4581cfb71e426"}]},"branch":"refs/heads/master"},"c3c26c8304058f78db9ced1c0b6516a45d2e00b1":{"kind":"REWORK","_number":2,"created":"2020-10-15 14:13:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/2"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 14:13:24.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3c26c8304058f78db9ced1c0b6516a45d2e00b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3c26c8304058f78db9ced1c0b6516a45d2e00b1"}]},"branch":"refs/heads/master"},"a58541b37a4d03ca5e221debc126579224809c98":{"kind":"REWORK","_number":3,"created":"2020-10-15 14:14:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/3"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 14:13:59.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a58541b37a4d03ca5e221debc126579224809c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a58541b37a4d03ca5e221debc126579224809c98"}]},"branch":"refs/heads/master"},"f0e022c9c6077e12fefa17f1b531eb60c6d1ce84":{"kind":"REWORK","_number":4,"created":"2020-10-15 14:56:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/4"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 14:56:13.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e022c9c6077e12fefa17f1b531eb60c6d1ce84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e022c9c6077e12fefa17f1b531eb60c6d1ce84"}]},"branch":"refs/heads/master"},"e9aa45b6877beb73e5ae76cf0ee7eca985e2c22d":{"kind":"REWORK","_number":5,"created":"2020-10-16 09:36:35.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/5"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-16 09:36:23.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9aa45b6877beb73e5ae76cf0ee7eca985e2c22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9aa45b6877beb73e5ae76cf0ee7eca985e2c22d"}]},"branch":"refs/heads/master"},"6971cfd638fe37a8f8190aa4e9d0fab1330946ef":{"kind":"REWORK","_number":6,"created":"2020-10-16 09:41:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/6"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-16 09:40:55.000000000","tz":120},"subject":"Managing the review inbox by serie","message":"Managing the review inbox by serie\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6971cfd638fe37a8f8190aa4e9d0fab1330946ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6971cfd638fe37a8f8190aa4e9d0fab1330946ef"}]},"branch":"refs/heads/master"},"be85d405e0bcec5d23af6126f46aa188621ccf2c":{"kind":"REWORK","_number":7,"created":"2020-10-22 14:37:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/7"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-22 14:37:18.000000000","tz":120},"subject":"Managing the review inbox by series","message":"Managing the review inbox by series\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be85d405e0bcec5d23af6126f46aa188621ccf2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be85d405e0bcec5d23af6126f46aa188621ccf2c"}]},"branch":"refs/heads/master"},"64e5906f9693de8ee192d9d6e497a0660edfa3e6":{"kind":"REWORK","_number":8,"created":"2020-10-22 14:50:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/8"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-22 14:50:05.000000000","tz":120},"subject":"Managing the review inbox by series","message":"Managing the review inbox by series\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64e5906f9693de8ee192d9d6e497a0660edfa3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64e5906f9693de8ee192d9d6e497a0660edfa3e6"}]},"branch":"refs/heads/master"},"1788c9edce159fedf7c3cd02a0501c749c05f6f0":{"kind":"REWORK","_number":9,"created":"2020-10-22 15:55:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/9"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-22 15:55:12.000000000","tz":120},"subject":"Managing the review inbox by series","message":"Managing the review inbox by series\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1788c9edce159fedf7c3cd02a0501c749c05f6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1788c9edce159fedf7c3cd02a0501c749c05f6f0"}]},"branch":"refs/heads/master"},"af777a9034e814622d34be96a374a5be7bb9b45f":{"kind":"REWORK","_number":10,"created":"2020-10-22 15:56:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/758378/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/758378/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/758378/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/758378/10"}}},"commit":{"parents":[{"commit":"dee725fa084beea849ef99bb1619c162a2e07a65","subject":"Add latest release note landing pages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dee725fa084beea849ef99bb1619c162a2e07a65"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-15 09:37:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-22 15:56:34.000000000","tz":120},"subject":"Managing the review inbox by series","message":"Managing the review inbox by series\n\nOfficially providing the release review inbox in the reviewer guide\nand consider its update in our process\n\nChange-Id: I7b040c5a5ef4f2dd4868b62c1498b5f2a3d8b827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af777a9034e814622d34be96a374a5be7bb9b45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/af777a9034e814622d34be96a374a5be7bb9b45f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
