)]}'
{"id":"openstack%2Freleases~758990","triplet_id":"openstack%2Freleases~master~I53aeb3211bb3251a3278472a514a39afe825cdd2","project":"openstack/releases","branch":"master","topic":"em-eol","hashtags":[],"change_id":"I53aeb3211bb3251a3278472a514a39afe825cdd2","subject":"Adding a tool to track project who need to drop eol branches","status":"MERGED","created":"2020-10-21 10:10:54.000000000","updated":"2020-11-12 12:54:33.000000000","submitted":"2020-11-12 12:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758990-1605185673053-1ed5524d","meta_rev_id":"7b7aa91fca629c8b0addf2d9be2fec880df446fe","_number":758990,"virtual_id_number":758990,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 12:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-12 10:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-12 10:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-11-12 10:57:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-12 12:40:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-01 10:33:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-11-11 15:00:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-12 12:40:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-11-12 12:54:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2274e7c6632b44b12d3f02de337538d1e7311c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-21 10:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edcd2ed40cda0575bce0422de9d208364edc0cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 10:38:32.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e35ed2d23df4e4db6791e0d459bcb7c : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c60c1220098447f681e9e8a70c4c5bac : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2794e4dea0cd4e98aa0ceea988382ad0 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7791cbc26e8b4df59216341463ba4009 : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29510c0fc996404cb87064623a72721b : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/146232f397f34fa3987a59757064f5ae : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/148a9ed7e3c042bbb3347fc1b403903b : SUCCESS in 6m 31s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c50a7ee5b004402eb1e52b6cebc4c51f : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7acecf4cfddcf999b569693a6384aabcab2bcb7f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 14:27:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"424e535334a4abebf66fff8b701efb75fff79747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 14:52:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4385560d58ee424394f9186f74a3e3ca : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c613fb1c1de64f0e9c9be969f6b5f83c : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c39dcee8573467cafc4d833304169b0 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f499809827c344bf9a657e3ca4570165 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b63ea95d5bf34896be4ae1f8180de6f5 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5cd6db3b918b4c48bff506e764d807d3 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6e27eec10474d1a8da8d8a6f821abfd : SUCCESS in 6m 48s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/84d6a1abcdc04bebbb136c6659c96828 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"0d872527cdf536fced421e8fd3dd66ddbd666c5f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-27 10:46:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"438b3cb00f31063e41866a444613050b665f9217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:16:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8dcb9845744545bf53eb6fe266b495 : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da8a99a1d04d4e1680cd2bd9f90ec863 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88835beb4fbf4997bf3c0a4a815cc19f : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd9d66fd41d44d89937e394f63c8b11 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce8188a5f4cf4518a905437229925689 : SUCCESS in 3m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/279d97f90be44778ae3dd2d900335cdd : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7d174d966284e04bd6ff6d60e744274 : SUCCESS in 4m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5745d2ebb14e4de2991ed6ba3fc23cca : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"63983a893934d76cef1082bb5e0929419cba1ace","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-29 16:32:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8574e4e049d5f5dc00f8d0d692dd38eab1d977e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 16:53:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf071df3d0af4d7e914087799f446a3e : FAILURE in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b73aeaa5531c4a62974babddbd753771 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/204ee9ba98d64815847e001fec89ea20 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3732e6036c41e0baa169b4532a78f2 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2832bb4a34e47eda8db46085ba1278d : SUCCESS in 7m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/73c0c7eb6f344e4cac0e7ed5f9ad1a74 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee61a4d44d2c4fc391d2b34e0a62c4a5 : SUCCESS in 5m 55s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fec716ded8e54453b65fee3ed80c1f92 : SUCCESS in 4m 25s\n\nWarning:\n  Comments left for invalid file openstack_releases/cmds/list_em_series.py","accounts_in_message":[],"_revision_number":4},{"id":"3daa87328503c35b9d8c0c359828a74f5cade946","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 08:32:00.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"bcc14697384d6745a7d650ca9793b934047e7f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 08:48:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ddee20c113487d9460aaaf76f34812 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12fa56e49f1e4d5fa51d6eb7c98cbce3 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b578407a845f4d9099d5f2ae1b1dfeff : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a291ffbeee4870a34aff2401961ab9 : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5ab81ebd15c452cbe21a242d2685c4e : SUCCESS in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1426b88854fd4ca7a73fcd6c8a052c86 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42fea986acb54c8ea59c8b4795c050c0 : SUCCESS in 4m 13s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a92662fbc71e453aa5531de3cac50b2c : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"fdee6d0c46a659cfeb360e5703f717388f6ab14f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 08:58:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"899e069c21ca72051c6724c679efe46342624191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 09:13:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84642b13652a436bb905cfa65129f67c : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dae66b4054947b6b6baf2dde28d76df : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860bfd309f65484697e0c6e87826f712 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9fafff045c64fef8bc58fc2e8a6b8f5 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de39cd62e4ee4c7ca202d634928f5ffe : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/52e0521cbc3546ce8a37f7887254fd8c : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a831f42fa0ac43ee87504f616a891a68 : SUCCESS in 5m 02s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/56ea0bd509f6483c92c8e85263f488aa : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"539d5447afcc60e644860622fedd151ca29d80ad","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-10-30 13:49:40.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nThe script worked for me locally, however I couldn\u0027t yet confirm that the output is fine as it seems there\u0027s not a single deleted branch that is EOL\u0027d in ocata/pike/queens/rocky. Is that the case really, or does somehow the script missing such branches? Do you have an example branch that is deleted?","accounts_in_message":[],"_revision_number":6},{"id":"3b7a1abc82fa0fb9074ed6ad3d5b1c1cd25d5dff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-30 15:21:20.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fa3f40e1634c65a6b0850661a4f3d27dc8cb9719","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 16:43:55.000000000","message":"Patch Set 6:\n\n\u003e (2 comments)\n \u003e \n \u003e The script worked for me locally, however I couldn\u0027t yet confirm\n \u003e that the output is fine as it seems there\u0027s not a single deleted\n \u003e branch that is EOL\u0027d in ocata/pike/queens/rocky. Is that the case\n \u003e really, or does somehow the script missing such branches? Do you\n \u003e have an example branch that is deleted?\n\nAFAIK (2 years at work on openstack) I never seen EM EOL branches deletion, only ocata recently asked for that, so I don\u0027t think we have some history on that.","accounts_in_message":[],"_revision_number":6},{"id":"8e7149955bd2558ecb18ae1e66cba125d78a2821","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 16:46:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cc508b46cd0374e12393f11c070b93210a509e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:11:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6553dba43534e0a9230657b596b20d5 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89856f0e5d424628952c621fdc164028 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20be298fdd9f4145af9c4d3e7aa2bf54 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ade371f092482b80ad77d3d727c9d3 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fa4b4f61c394ae7b2783e0a6c23fae5 : SUCCESS in 4m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e03c86f8266844eeaeb4364627d1888f : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc7de3ed4b9c43d599e000775167712b : SUCCESS in 4m 59s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/03ba99e612c346dea9e8746b23e33664 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":7},{"id":"9b78cec718c11efc2a4efc2ee25ef5bbdde5301c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-01 10:33:50.000000000","message":"Patch Set 7: Code-Review+1\n\n\u003e only ocata recently asked for that\n\nYou mean Octavia, don\u0027t you? :)\nAnyway, the tool looks good, thanks for working on this Hervé! :)","accounts_in_message":[],"_revision_number":7},{"id":"100fee23f828a593d0db5b37b5737da32fd7c1b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-02 08:52:25.000000000","message":"Patch Set 7:\n\n\u003e \u003e only ocata recently asked for that\n \u003e \n \u003e You mean Octavia, don\u0027t you? :)\n \u003e Anyway, the tool looks good, thanks for working on this Hervé! :)\n\nYes sorry I meant Octavia.","accounts_in_message":[],"_revision_number":7},{"id":"da3b32b05a5798fa74c0542ad900cac703cafdf4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-02 14:53:49.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ea283825c81c6f366e2296a37bcc8c9db66f2f19","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-11 15:00:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"351d4040aadebc8998c75fc45db9ea7479fd042c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-12 10:57:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"7109bdd32d6f7f77fa01644bad7a01d698f99e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 11:15:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade478e9874348d08a643137a7506547 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c99c6a2b33d46428980bdf202cd86a0 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4630c16a82004ab18ce04fc740eeea09 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d645c1f61f5405bbcac43a6d08d465d : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cded84294604d218299c54ae45796c2 : SUCCESS in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2869c12fd1a244f886699fd922190074 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28095e08cf4f484c87fd83775e113322 : SUCCESS in 4m 40s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b2576e5830cc4ac08f5a04fd6d702a1e : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":8},{"id":"79019a5d3e7e2378b0d712c482555fc86cfc1db1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-12 12:40:05.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"36117d82274c35c92c1b2b25054b6aa72a958174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:40:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d63d1ec44e603818cf872c68d41e90791cbd862a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:54:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b8ae83289c4a0badb2875f626b3679 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0c24050e4854d82b56d6c0e2525ef31 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b88c469ba154936b8beeef7122f33cc : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c383af8cf854dfcb3f5e11205713b69 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/887eb52f834747ea936e728c98fffedc : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16037151867a42b19097eb92c369b133 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":8},{"id":"ed0e932c890571d4008874f7ab39c08affa3ec51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:54:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"003eb35717b5c81e9a6db696ad8731981d45b8ab","revisions":{"a58beb18dcee8bec4eaa0201c12f8fcee4dc66bd":{"kind":"REWORK","_number":1,"created":"2020-10-21 10:10:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/1"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 18:44:39.000000000","tz":120},"subject":"[WIP] Adding a tool to track project who need to drop eol branches","message":"[WIP] Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nThese changes are split in tree parts:\n1. script added to track these stale branches\n2. existing code adapted to list deliverables who matches eol in a given serie\n3. short term process added to regularly check this existing eol and\nnotify the infra to ask them to remove the stale branches. Mostly to\nvaccum our environment during Wallaby and ensure to retrieve a clean env\nduring the next cycle.\n4. long term process added to ensure that the release team will notify the\ninfra team as soon as a branch is tagged as eol.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a58beb18dcee8bec4eaa0201c12f8fcee4dc66bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a58beb18dcee8bec4eaa0201c12f8fcee4dc66bd"}]},"branch":"refs/heads/master"},"21d8e6841ed97a18ca491078dc6c26a9dd0a8b1e":{"kind":"REWORK","_number":2,"created":"2020-10-22 14:27:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/2"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-22 14:27:12.000000000","tz":120},"subject":"[WIP] Adding a tool to track project who need to drop eol branches","message":"[WIP] Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nThese changes are split in tree parts:\n1. script added to track these stale branches\n2. existing code adapted to list deliverables who matches eol in a given serie\n3. short term process added to regularly check this existing eol and\nnotify the infra to ask them to remove the stale branches. Mostly to\nvaccum our environment during Wallaby and ensure to retrieve a clean env\nduring the next cycle.\n4. long term process added to ensure that the release team will notify the\ninfra team as soon as a branch is tagged as eol.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21d8e6841ed97a18ca491078dc6c26a9dd0a8b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21d8e6841ed97a18ca491078dc6c26a9dd0a8b1e"}]},"branch":"refs/heads/master"},"d5125cbc65eeb62651f9569663f3173db3188a38":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-27 10:46:05.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/3"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-27 10:42:57.000000000","tz":60},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5125cbc65eeb62651f9569663f3173db3188a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5125cbc65eeb62651f9569663f3173db3188a38"}]},"branch":"refs/heads/master"},"5f0cd49682bd81038124b9c04c02d9d6e8d02669":{"kind":"REWORK","_number":4,"created":"2020-10-29 16:32:46.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/4"}}},"commit":{"parents":[{"commit":"f5ed0f0bfac29580a67d60c704380ab38d462512","subject":"add tooling to list em series from data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5ed0f0bfac29580a67d60c704380ab38d462512"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-29 16:32:31.000000000","tz":60},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f0cd49682bd81038124b9c04c02d9d6e8d02669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f0cd49682bd81038124b9c04c02d9d6e8d02669"}]},"branch":"refs/heads/master"},"720e0630153c10e692da2640f21a4c92b990773e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-30 08:32:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/5"}}},"commit":{"parents":[{"commit":"89d4417ca14bf563167ab7427c761f2161a9ffa5","subject":"add tooling to list em series from data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89d4417ca14bf563167ab7427c761f2161a9ffa5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-30 08:31:01.000000000","tz":60},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/720e0630153c10e692da2640f21a4c92b990773e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/720e0630153c10e692da2640f21a4c92b990773e"}]},"branch":"refs/heads/master"},"7d277a84ebd12ed59d22469cdad09294efeca1cc":{"kind":"REWORK","_number":6,"created":"2020-10-30 08:58:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/6"}}},"commit":{"parents":[{"commit":"89d4417ca14bf563167ab7427c761f2161a9ffa5","subject":"add tooling to list em series from data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89d4417ca14bf563167ab7427c761f2161a9ffa5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-30 08:58:42.000000000","tz":60},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d277a84ebd12ed59d22469cdad09294efeca1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d277a84ebd12ed59d22469cdad09294efeca1cc"}]},"branch":"refs/heads/master"},"c48a52f9ff0f690bd74ea33d50ad4482ccdeaa6d":{"kind":"REWORK","_number":7,"created":"2020-10-30 16:46:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/758990/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/7"}}},"commit":{"parents":[{"commit":"89d4417ca14bf563167ab7427c761f2161a9ffa5","subject":"add tooling to list em series from data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89d4417ca14bf563167ab7427c761f2161a9ffa5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-30 16:46:03.000000000","tz":60},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c48a52f9ff0f690bd74ea33d50ad4482ccdeaa6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c48a52f9ff0f690bd74ea33d50ad4482ccdeaa6d"}]},"branch":"refs/heads/master"},"003eb35717b5c81e9a6db696ad8731981d45b8ab":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-11-12 10:57:39.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/90/758990/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/758990/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/758990/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/758990/8"}}},"commit":{"parents":[{"commit":"9289284e33ee76c331188cec4adf7b7cbc4124ed","subject":"[qinling] Transition Stein to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9289284e33ee76c331188cec4adf7b7cbc4124ed"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-10-20 14:27:42.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-12 10:57:39.000000000","tz":0},"subject":"Adding a tool to track project who need to drop eol branches","message":"Adding a tool to track project who need to drop eol branches\n\nSince transition from the EOL model to the EM model eol branches\nare no longer removed and this step is no longer in the release team\nprocess.\n\nKeep stale branches can be an issue in some situation especially with\nzuul and our gates. To avoid this situation the release team propose\nto reintroduce regular checks to ensure that we remove stale branches\nthat have been tagged eol previously.\n\nAs discussed during our previous meetings soon it will be possible to design\na new job and to trigger it to remove eol branches automatically.\nThis will possible when the infra would have been updated and when all\nthe needed pieces will be in place.\n\nChange-Id: I53aeb3211bb3251a3278472a514a39afe825cdd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/003eb35717b5c81e9a6db696ad8731981d45b8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/003eb35717b5c81e9a6db696ad8731981d45b8ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
