)]}'
{"id":"openstack%2Freleases~759006","triplet_id":"openstack%2Freleases~master~I14487be37e170a3981505962062120f5d17530b4","project":"openstack/releases","branch":"master","topic":"bug/1234567","hashtags":[],"change_id":"I14487be37e170a3981505962062120f5d17530b4","subject":"Add release notes links for Victoria release","status":"ABANDONED","created":"2020-10-21 11:53:13.000000000","updated":"2020-10-21 12:43:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af421804bcb7164140d26aadade9f79b0ea2e2f6","_number":759006,"virtual_id_number":759006,"owner":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-21 12:06:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"date":"2020-10-21 12:20:33.000000000","_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 12:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-21 12:20:33.000000000","updated_by":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"reviewer":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"state":"REVIEWER"}],"messages":[{"id":"07777d3db0a736a9ff1329b36eaa667bbf6b3f0e","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-21 11:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"135536af50dc32250bd4fd8f1398e5c4ab516b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 12:06:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b166c86cb06d46b78645ce5d983abf92 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c4e65078fff44809d19b191d8a17797 : FAILURE in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1396d543b8b64bd395c4a79a7cfd7b21 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0774a1b23a9c6ff6fad3656886e7114023ef5fd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 12:43:44.000000000","message":"Abandoned\n\nThese should only be done for projects that actually have release notes published. The tools/add_release_note_links.sh tool should be used to do this. And it should all be in one file.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"033b1c8542fb989fd1dcd4dd71182a045e63d7ca","revisions":{"033b1c8542fb989fd1dcd4dd71182a045e63d7ca":{"kind":"REWORK","_number":1,"created":"2020-10-21 11:53:13.000000000","uploader":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"ref":"refs/changes/06/759006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/759006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/759006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/759006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/759006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/759006/1"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-21 11:47:24.000000000","tz":480},"committer":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-21 11:47:24.000000000","tz":480},"subject":"Add release notes links for Victoria release","message":"Add release notes links for Victoria release\n\nChange-Id: I14487be37e170a3981505962062120f5d17530b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/033b1c8542fb989fd1dcd4dd71182a045e63d7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/033b1c8542fb989fd1dcd4dd71182a045e63d7ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
