)]}'
{"id":"openstack%2Freleases~759340","triplet_id":"openstack%2Freleases~master~Ide7f393814b6926f47abd3c0b3f0dcdf9fb8b759","project":"openstack/releases","branch":"master","topic":"ug/444","hashtags":[],"change_id":"Ide7f393814b6926f47abd3c0b3f0dcdf9fb8b759","subject":"New bugfix release for oslo.messaging","status":"ABANDONED","created":"2020-10-23 01:40:55.000000000","updated":"2020-10-26 07:17:23.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ca855e7420a06fcc390f3b258c84df5466620b1","_number":759340,"virtual_id_number":759340,"owner":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-23 02:27:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"date":"2020-10-26 06:49:05.000000000","_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-2,"date":"2020-10-23 07:34:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-23 11:20:07.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 02:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 07:34:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-23 11:20:07.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"2f6e8f4f9213b6134532aeaa252fbbedf3903543","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-23 01:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b135205073a75244391b2cda6c2dc8475a801f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 01:56:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcbaad65892420fa314299a8185fa9e : FAILURE in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/906263c32ba24746ab38f4b60dfe12bb : FAILURE in 5m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6845c6dcf4b346a18527c3fbf06304d3 : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"15253d8008b22845b6fa1448deaadad95ebf289e","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-23 01:59:30.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcbaad65892420fa314299a8185fa9e\n \u003e : FAILURE in 5m 36s\n \u003e - openstack-tox-validate https://zuul.opendev.org/t/openstack/build/906263c32ba24746ab38f4b60dfe12bb\n \u003e : FAILURE in 5m 01s\n \u003e - releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6845c6dcf4b346a18527c3fbf06304d3\n \u003e : FAILURE in 5m 40s\n\nthank you","accounts_in_message":[],"_revision_number":1},{"id":"7945ca08060cd54e1b272c680660d75eb53b324a","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-23 02:01:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"166b7a311857d7380face1db407d28b372278d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 02:27:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fca709197174846b0849f861041f885 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4de85fe6cb6d442895b96bc119a17e99 : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/83a4ecd89b64450b8cda222f54768db8 : RETRY_LIMIT in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"76e6924eecdd041dd822d7f2f1766d62ec1efaa3","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-23 02:29:25.000000000","message":"Patch Set 2:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fca709197174846b0849f861041f885\n \u003e : SUCCESS in 7m 48s\n \u003e - openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4de85fe6cb6d442895b96bc119a17e99\n \u003e : SUCCESS in 5m 47s\n \u003e - releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/83a4ecd89b64450b8cda222f54768db8\n \u003e : RETRY_LIMIT in 3m 05s\n\nthank you","accounts_in_message":[],"_revision_number":2},{"id":"0e39fa7bd5a4e40a71316f42ff949a7e2ad681ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-23 07:34:14.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3306811c3b224a389513bf3ff39ac84d08ffd486","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-26 06:49:05.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nthank you","accounts_in_message":[],"_revision_number":2},{"id":"c9d1bae1fb7a4601331c5c2f5c0da9ac20f5d192","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-26 07:17:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aec2d720d58167e16c2a6402767a3afa0b38a361","revisions":{"9d7de3f8facdbe67cf8294d49b572dc5a7d66b4c":{"kind":"REWORK","_number":1,"created":"2020-10-23 01:40:55.000000000","uploader":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"ref":"refs/changes/40/759340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/759340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/759340/1"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-23 01:42:13.000000000","tz":480},"committer":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-23 01:42:13.000000000","tz":480},"subject":"New bugfix release for oslo.messaging","message":"New bugfix release for oslo.messaging\n\nChange-Id: Ide7f393814b6926f47abd3c0b3f0dcdf9fb8b759\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d7de3f8facdbe67cf8294d49b572dc5a7d66b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d7de3f8facdbe67cf8294d49b572dc5a7d66b4c"}]},"branch":"refs/heads/master"},"aec2d720d58167e16c2a6402767a3afa0b38a361":{"kind":"REWORK","_number":2,"created":"2020-10-23 02:01:17.000000000","uploader":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"ref":"refs/changes/40/759340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/759340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/759340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/759340/2"}}},"commit":{"parents":[{"commit":"87c71eebd46782fcfdc81c3f53e823e4de6a3754","subject":"Merge \"sahara: tag pike as EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87c71eebd46782fcfdc81c3f53e823e4de6a3754"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-23 01:42:13.000000000","tz":480},"committer":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-23 02:01:13.000000000","tz":0},"subject":"New bugfix release for oslo.messaging","message":"New bugfix release for oslo.messaging\n\nChange-Id: Ide7f393814b6926f47abd3c0b3f0dcdf9fb8b759\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aec2d720d58167e16c2a6402767a3afa0b38a361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aec2d720d58167e16c2a6402767a3afa0b38a361"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
