)]}'
{"id":"openstack%2Freleases~761718","triplet_id":"openstack%2Freleases~master~I57a3f0fa57e5418229a72942530a9244ea4906e2","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I57a3f0fa57e5418229a72942530a9244ea4906e2","subject":"Release cliff 3.5.0 for Wallaby","status":"MERGED","created":"2020-11-06 08:51:04.000000000","updated":"2020-11-11 15:17:17.000000000","submitted":"2020-11-11 15:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761718-1605107837292-28503777","meta_rev_id":"dd2e9e5cd520f42e6c54f38c011189a5c223f00d","_number":761718,"virtual_id_number":761718,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 15:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-11 15:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-06 10:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-11-06 10:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-11 15:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-11-06 09:28:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-06 10:05:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-06 10:20:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-11-11 15:01:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-11 15:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d51065c320fe60c615806d88e3156a89e74c742a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-11-06 08:51:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ca03441f292ecbabf7947462a9b60c8dacfa47d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 08:52:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"704d40d2da45fae56f41a4dfd886855278e31634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 09:06:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c99e0496d446119cf7317b200fae3c : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/438902637fad499eb79f5acb5421d4d3 : FAILURE in 3m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cd70d88c233943788fa0767bebf50dd9 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a76079af56c8ae5a815f2fcc808a94c0d236943b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-11-06 09:26:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b826b25b40b897dca0da24ab9f1abc623900a4da","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 09:28:16.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"90379c48e8fbd66313cb86db507d323e8b03cb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 09:35:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c92ddd1cc94c0ab1ad78a7cd4b0482 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d24cbbb89fd49d1837d477c1a9b3ded : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1effe741510a4775845f3058404fc290 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"113d32a3a78945e207c0c8eda45a4320890f6afb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-06 10:05:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"590215e1b7a16c04dcbe87489f920e5cdc1a3010","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-06 10:20:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e0489838808aa8e139ac9769da848743f6ec339a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-11 15:01:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"967204a6ac73fb1891c3b960facab864d01cbc3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 15:02:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fcbe55d1d14cf7945399592901f5e864a96dc217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 15:17:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0663e4d1c2eb47f4aa44b7cde3dd5661 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51717aa096d14e12aea6e951daa08495 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f0f540d92754ec1ab5eb35d33feed4f : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"dd2e9e5cd520f42e6c54f38c011189a5c223f00d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 15:17:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ef2c57626c7e6be1173edeff4a30064f7d32451","revisions":{"6ff728f1c56a4812f281178c2dea5867d99e7aec":{"kind":"REWORK","_number":1,"created":"2020-11-06 08:51:04.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/18/761718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/761718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/761718/1"}}},"commit":{"parents":[{"commit":"e8cd9543de2ddc89084466e6bdc56a7ba5e0f4b7","subject":"Merge \"add tooling to list em series from data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8cd9543de2ddc89084466e6bdc56a7ba5e0f4b7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-11-06 08:50:43.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-11-06 08:50:43.000000000","tz":60},"subject":"Release cliff 3.5.0 for Wallaby","message":"Release cliff 3.5.0 for Wallaby\n\nChange-Id: I57a3f0fa57e5418229a72942530a9244ea4906e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ff728f1c56a4812f281178c2dea5867d99e7aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ff728f1c56a4812f281178c2dea5867d99e7aec"}]},"branch":"refs/heads/master"},"2ef2c57626c7e6be1173edeff4a30064f7d32451":{"kind":"REWORK","_number":2,"created":"2020-11-06 09:26:58.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/18/761718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/761718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/761718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/761718/2"}}},"commit":{"parents":[{"commit":"e8cd9543de2ddc89084466e6bdc56a7ba5e0f4b7","subject":"Merge \"add tooling to list em series from data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8cd9543de2ddc89084466e6bdc56a7ba5e0f4b7"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-11-06 08:50:43.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-11-06 09:26:45.000000000","tz":60},"subject":"Release cliff 3.5.0 for Wallaby","message":"Release cliff 3.5.0 for Wallaby\n\nChange-Id: I57a3f0fa57e5418229a72942530a9244ea4906e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ef2c57626c7e6be1173edeff4a30064f7d32451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ef2c57626c7e6be1173edeff4a30064f7d32451"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
