)]}'
{"id":"openstack%2Freleases~762300","triplet_id":"openstack%2Freleases~master~I21639735b30cbaea11fc9a2a5b305539db823e11","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I21639735b30cbaea11fc9a2a5b305539db823e11","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"ABANDONED","created":"2020-11-11 07:11:52.000000000","updated":"2020-11-11 12:04:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44f536b7d3477ca0f34479199ad137b17aba4eee","_number":762300,"virtual_id_number":762300,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-11 08:17:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2020-11-11 10:41:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 08:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 10:41:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d8abc570394c56d5388e07652ea54935b2ea0471","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-11 07:11:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07c8140a5f127fdfead848d90bb29dc73fd683f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 08:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2630db0c504a658b51a0648beaa032 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5595d816677246efa27dd12ffeb7a8ed : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a60d607806e644ccb0dd84610f771fa9 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1875c02ee15948319b78925cd16f6432 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31c24f18e46c4fe790dc5811a2671e80 : SUCCESS in 4m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dccce68a71314ce789f2f09e26025949 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/717ec6ae0d80441e8e51611370558fb6 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"62ec25de307fb1d07a42e1bd46ec5d51b34641f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-11 10:41:45.000000000","message":"Patch Set 1: Code-Review-1\n\nSee https://review.opendev.org/#/c/762291/","accounts_in_message":[],"_revision_number":1},{"id":"44f536b7d3477ca0f34479199ad137b17aba4eee","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-11 12:04:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e8c76a9b7955d1e5ddb6efed1cedb107c97718c","revisions":{"2e8c76a9b7955d1e5ddb6efed1cedb107c97718c":{"kind":"REWORK","_number":1,"created":"2020-11-11 07:11:52.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/00/762300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/762300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/762300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/762300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/762300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/762300/1"}}},"commit":{"parents":[{"commit":"f0164d86a9123e8c05403a2b3cfd9888d0eb4b46","subject":"Merge \"Release OpenStack-Ansible Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0164d86a9123e8c05403a2b3cfd9888d0eb4b46"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 07:08:14.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 07:11:46.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nChange-Id: I21639735b30cbaea11fc9a2a5b305539db823e11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e8c76a9b7955d1e5ddb6efed1cedb107c97718c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e8c76a9b7955d1e5ddb6efed1cedb107c97718c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
