)]}'
{"id":"openstack%2Freleases~762395","triplet_id":"openstack%2Freleases~master~Iff6b798225909d46f8c49927e8fad1eab3ebdfe0","project":"openstack/releases","branch":"master","topic":"stein-em","hashtags":[],"change_id":"Iff6b798225909d46f8c49927e8fad1eab3ebdfe0","subject":"[manila] Transition Stein to EM","status":"MERGED","created":"2020-11-11 19:24:45.000000000","updated":"2020-11-19 17:24:08.000000000","submitted":"2020-11-19 17:24:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762395-1605806648504-4f1731f4","meta_rev_id":"fe90996adb0f91c659b21e8cb4d9de1678aefa72","_number":762395,"virtual_id_number":762395,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-11-19 13:48:56.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 17:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-11-19 14:38:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-11-19 17:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-11-19 14:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-11-19 17:05:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-11-19 14:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 20:20:51.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-11-19 14:31:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-19 14:38:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-11-19 17:05:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-11-19 17:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70607546041432b17facc44cfa54d4c30cf471cf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-11 19:24:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"219f7773c3485061aadc8a29212640010d2193c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 19:37:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe48f978b6a41c4b2ed3b454aec9ffb : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdb6c6be50ea48a38959f147ffae6f24 : FAILURE in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3aba64bb8cd943269e4fad593295bb7e : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"16e2c30ec896af824d96947e65ce1e4f384db2dc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-11 22:49:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2b98b3c469645e06ca0988e3881224ff18a09c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 22:55:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19a8e4410e94420a2fe7d06bd4bcd31 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/880c4094d0bd44ecb5dd119319e3dcae : SUCCESS in 4m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/63488189888740e18a213a1e15b4ca87 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9ef03559869cd78a8554f9c997ec85407cf92e0c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-11 23:42:32.000000000","message":"Patch Set 2: Code-Review-1\n\nThank you for this change - there\u0027s a patch in progress - would be good to have it merge: https://review.opendev.org/762110/","accounts_in_message":[],"_revision_number":2},{"id":"6ebdc69dc17b504ebe8b095bf25879e3fb863ba2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-12 02:47:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f30ea08989f9cb5feee8316ef48c1a0bbdad25b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 02:48:55.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"b525dab0afa148b2a0cdc9997c33e6ee962d3d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 02:59:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bc685ba68641c9b7d0dad72f460453 : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e0abeea7ebd4fd8ba72f2483bc53f98 : FAILURE in 5m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6e4e373b4b9a4a4ebd88a304a4f99cae : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"564eb28a0b6b13cec1906b43ccd20304303970e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-12 06:12:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3d897bc31bcddf4a1c744d65e30fa2c8e8b42f1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 06:14:05.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"a04896927f397f3da7b71a3bf13a233c6f6dab8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 06:32:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4985fc08fe6048a3a2732c8835ca26c7 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bff1d82afa541d78b3cd00b8727e926 : FAILURE in 9m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b4712b09c63a4fab851c6a95c77dcf37 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0265a2cae1a62b846f215b280afa66fc4da01aec","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-12 09:25:18.000000000","message":"Topic changed from stable-em to stein-em","accounts_in_message":[],"_revision_number":4},{"id":"1654e5ecaf0c4ec5b4914565bc68da41ce7f4bce","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-12 09:37:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"02a4130e75b6c5ee5e777d2fabfc7beef73f809f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-12 09:41:06.000000000","message":"Patch Set 5:\n\nThanks @Goutham, I\u0027ve separated the patch to \n- a release patch and\n- an EM tagging patch\nfor the sake of simplicity (and I think validator likes that better).","accounts_in_message":[],"_revision_number":5},{"id":"6982e7f571330b186cf7b352ff8977cc8ac10213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 09:53:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f90eaaa2cdc44e9a50d169a6c961c87 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39278d874521489e876ae5c15deb3aa3 : FAILURE in 7m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/719c8ea108294dfdb629b5fef4789e41 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"abfafaddf67d65200af6a5429eb5e0f90c290738","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-12 17:40:37.000000000","message":"Patch Set 5:\n\n\u003e Thanks @Goutham, I\u0027ve separated the patch to \n\n\u003e a release patch and\n\u003e an EM tagging patch\n\u003e for the sake of simplicity (and I think validator likes that better).\n\nThank you! I agree, i suppose the validator will pass once we merge the release patch?","accounts_in_message":[],"_revision_number":5},{"id":"a944962392b42641f1756b5871e0bb26ac5950f8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-12 18:24:37.000000000","message":"Patch Set 5:\n\n\u003e \u003e Thanks @Goutham, I\u0027ve separated the patch to\n \u003e \n \u003e \u003e a release patch and\n \u003e \u003e an EM tagging patch\n \u003e \u003e for the sake of simplicity (and I think validator likes that\n \u003e better).\n \u003e \n \u003e Thank you! I agree, i suppose the validator will pass once we merge\n \u003e the release patch?\n\nYes, it will pass. :)","accounts_in_message":[],"_revision_number":5},{"id":"f014c43037d746239699541284bf0a3afeb78670","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-19 13:48:56.000000000","message":"Patch Set 5:\n\nrecheck - as the parent release patch has merged.","accounts_in_message":[],"_revision_number":5},{"id":"0aae28ee477be22d6c5ff3fcbb77a33e79760c3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:03:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef892c9978d467385b5fce74ceb421d : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/868f4bbbd51541848fbeebaa84c910f4 : SUCCESS in 4m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/12e8a4adf1af4eaf96de81503cb7c732 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":5},{"id":"4dca73447ffd0d7c6708db9ccd01bbf56df66d17","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 14:31:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"83ddb47dda162f0656cb4d8822ec44544cb7d4f3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 14:38:23.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":5},{"id":"ad5b51b47bff91c4de4d05933e7fd73ac32cd990","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:39:34.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"94eb26506e6ede3747b83c0227ff51911895f5f7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-19 17:05:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"27a860383701d065c2409b250ee8077c07599658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:06:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3782f2d0ae00c19f2deaa87c7bb63e10c994a232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:24:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8919d5e3c94b108ccd78eb50191248 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3612393ed66f4b8789041955335ffec0 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ce397d2e92449c78f747a80aa41115c : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":5},{"id":"fe90996adb0f91c659b21e8cb4d9de1678aefa72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:24:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a5752bfce7d5485484f62aa5af9590ae1944685","revisions":{"4be493dd9a0946c59c3a84f17c784cd8db859cfd":{"kind":"REWORK","_number":1,"created":"2020-11-11 19:24:45.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/762395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/762395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/762395/1"}}},"commit":{"parents":[{"commit":"83eff99844d0301e608369a876b11bf861998f20","subject":"[magnum] Transition Stein to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83eff99844d0301e608369a876b11bf861998f20"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"subject":"[manila] Transition Stein to EM","message":"[manila] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Iff6b798225909d46f8c49927e8fad1eab3ebdfe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4be493dd9a0946c59c3a84f17c784cd8db859cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4be493dd9a0946c59c3a84f17c784cd8db859cfd"}]},"branch":"refs/heads/master"},"7bdb0182994e2e210434b57c3568053463fb332b":{"kind":"REWORK","_number":2,"created":"2020-11-11 22:49:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/762395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/762395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/762395/2"}}},"commit":{"parents":[{"commit":"83eff99844d0301e608369a876b11bf861998f20","subject":"[magnum] Transition Stein to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83eff99844d0301e608369a876b11bf861998f20"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 22:48:44.000000000","tz":60},"subject":"[manila] Transition Stein to EM","message":"[manila] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Iff6b798225909d46f8c49927e8fad1eab3ebdfe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdb0182994e2e210434b57c3568053463fb332b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdb0182994e2e210434b57c3568053463fb332b"}]},"branch":"refs/heads/master"},"c00f0eda68898e685301233eebacd3494231c1ae":{"kind":"REWORK","_number":3,"created":"2020-11-12 02:47:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/95/762395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/762395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/762395/3"}}},"commit":{"parents":[{"commit":"83eff99844d0301e608369a876b11bf861998f20","subject":"[magnum] Transition Stein to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83eff99844d0301e608369a876b11bf861998f20"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-12 02:46:58.000000000","tz":-480},"subject":"[manila] Transition Stein to EM","message":"[manila] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nThis change creates the 8.1.4 release of manila,\nthat includes several bug-fixes.\n\nChange-Id: Iff6b798225909d46f8c49927e8fad1eab3ebdfe0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c00f0eda68898e685301233eebacd3494231c1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c00f0eda68898e685301233eebacd3494231c1ae"}]},"branch":"refs/heads/master"},"3ef29ea3f7e83648bb94427a1dd99fa62d6240b9":{"kind":"REWORK","_number":4,"created":"2020-11-12 06:12:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/95/762395/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/762395/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/762395/4"}}},"commit":{"parents":[{"commit":"83eff99844d0301e608369a876b11bf861998f20","subject":"[magnum] Transition Stein to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83eff99844d0301e608369a876b11bf861998f20"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-12 06:11:24.000000000","tz":-480},"subject":"[manila] Transition Stein to EM","message":"[manila] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nThis change creates the 8.1.4 release of manila,\nthat includes several bug-fixes.\n\nChange-Id: Iff6b798225909d46f8c49927e8fad1eab3ebdfe0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ef29ea3f7e83648bb94427a1dd99fa62d6240b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3ef29ea3f7e83648bb94427a1dd99fa62d6240b9"}]},"branch":"refs/heads/master"},"5a5752bfce7d5485484f62aa5af9590ae1944685":{"kind":"REWORK","_number":5,"created":"2020-11-12 09:37:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/762395/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/762395/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/762395/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/762395/5"}}},"commit":{"parents":[{"commit":"6ab1ae07ea88fb98a448076f173e48fe672b8405","subject":"Release manila 8.1.4 for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ab1ae07ea88fb98a448076f173e48fe672b8405"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:23:57.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-12 09:37:00.000000000","tz":60},"subject":"[manila] Transition Stein to EM","message":"[manila] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nChange-Id: Iff6b798225909d46f8c49927e8fad1eab3ebdfe0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a5752bfce7d5485484f62aa5af9590ae1944685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a5752bfce7d5485484f62aa5af9590ae1944685"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
