)]}'
{"id":"openstack%2Freleases~762397","triplet_id":"openstack%2Freleases~master~I732722b866234ca5d034272a18ca787a1106986c","project":"openstack/releases","branch":"master","topic":"stein-em-patch2","hashtags":[],"change_id":"I732722b866234ca5d034272a18ca787a1106986c","subject":"[masakari] Transition Stein to EM","status":"MERGED","created":"2020-11-11 19:28:25.000000000","updated":"2020-11-18 15:13:48.000000000","submitted":"2020-11-18 15:13:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762397-1605712428543-422172d5","meta_rev_id":"dc03715e40dcfbcd1de4052d16e17562e88ef1a2","_number":762397,"virtual_id_number":762397,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-18 15:13:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-18 14:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-17 15:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-11-17 15:28:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-18 14:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-11-17 15:29:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 15:28:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-17 15:30:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-18 14:55:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-18 15:13:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4afa8a4717998285421d3ce9c8fc2eb62035d258","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-11 19:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3615147f5cf8f7f4713a43eea4870bfae14c3401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 19:43:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c65f789a3b084521b75cd1916e1e285b : SUCCESS in 8m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ae769a9ef9b406f8e953633e46edcae : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/de2eb23697554cb599bfdb6243177166 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4bac55782d7b850dbf134482c84916349890ee82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 07:53:47.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for the reminder. I need to do some extra final releases.","accounts_in_message":[],"_revision_number":1},{"id":"36453c8424902feb43d296a6da08ee0552033478","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 17:23:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"814ced1973973539fbfed44699b8c90d5283809a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 17:24:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"224f81d48d883ec7e18a7de4ed5d2fbfe3a4adba","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 17:25:06.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9227b62465a27adb5370aabd8f0cde1d84ca3ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 17:34:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f320a5d29a4aca8768b3b94101cd43 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e69e3e6bcf7473c9bbc8348995a5e8c : FAILURE in 4m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/66ea4ef73eba498ebd7a6bc9fe1ddeb6 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"80a1abf609b6f99e1307068f6f624509ab403331","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 17:37:16.000000000","message":"Patch Set 2:\n\nto recheck after the previous merges and is acted upon","accounts_in_message":[],"_revision_number":2},{"id":"822c123d9b5c92f8b86e6453a364f24d61e8dab9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 18:55:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8bf3b82e4acbc75c0cd94d09454109f60b6d76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:08:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bde0f853f7744c3b253d16c22175345 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccd3399138a44d5c935c925362d81b5a : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4c6443ef93cf424faf1b7fcff48f4069 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"81ecb3616d404d5dd052dd632a5afff1a0fad45b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-13 10:48:12.000000000","message":"Patch Set 2: Workflow-1\n\nHolding this one by waiting to solve http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018733.html","accounts_in_message":[],"_revision_number":2},{"id":"9ca9ab91f551f847195aba406666e10cc509eac7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-13 12:47:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba7c51fbc070edab2ff84f5569a6f0cf5ed1482b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-13 12:47:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"11cd3cd248893209d8418f81802e00bb0c3efeb5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 12:48:43.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"4563a9789cc4f86f46e6bab64d5730d9608c421b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 12:58:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4544595538cd4a19948bdb1f87f60151 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36740d2e69de4f0fb73d0428d51335c4 : FAILURE in 4m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5edceea17090488c95572d0a54f81bea : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a546f8e5839f1af0d5333dac8514b1ba60287a4e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-13 12:58:59.000000000","message":"Patch Set 3:\n\nto recheck after the previous merges and is acted upon","accounts_in_message":[],"_revision_number":3},{"id":"8ea42ffed40a6971a683ec8bcb6a13a0d6213333","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-17 15:28:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5aaf6bdb5540b12dcdc2807ef49e50bea7810dc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-17 15:28:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8ec0b9044312bd12f9931dcdd7761a11e83c4d9f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:29:58.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"bb5115801d7d1f7e61f7ac1b66183066a1bc7e0c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-17 15:30:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6c764690673d54d1454861a40af874de133794b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:40:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb0e6ab18534b269d9bdc4f701dd671 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6c9d3000458400a9f44ceb9b9ab00f6 : SUCCESS in 5m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/54ba568b146f4a69aecf861990461bd9 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b316e659c76fdcc2514d66a31dcf882af3bb0d3b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-18 14:55:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"33a1ac0678319799868d11c4bf8ca5da29912394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 14:56:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"053669d8410945aa7b8a4d338bf834bfed500a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 15:13:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a1acfaa04f29f8803ee3ee18cc1db21419f232ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 15:13:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d0b216ace9461eadd8f22162204da5 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501c50279a35424fa5ccfbd25ccab165 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/399abb84099b4d36a95ce156e4339e45 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"51b9905be9b82f0d1be9712f7106403c7d928fae","revisions":{"673b2d8d6aca92c8a81dd1917b0ee7cec4c06145":{"kind":"REWORK","_number":1,"created":"2020-11-11 19:28:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/97/762397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/762397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/762397/1"}}},"commit":{"parents":[{"commit":"32499bc048a68f4ddc08a8af3d4b0d15d365413d","subject":"Merge \"Final Cinder team releases for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32499bc048a68f4ddc08a8af3d4b0d15d365413d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:27:46.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:27:46.000000000","tz":60},"subject":"[masakari] Transition Stein to EM","message":"[masakari] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I732722b866234ca5d034272a18ca787a1106986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/673b2d8d6aca92c8a81dd1917b0ee7cec4c06145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/673b2d8d6aca92c8a81dd1917b0ee7cec4c06145"}]},"branch":"refs/heads/master"},"95b49dfc56ab156212927bb5780faee1e743a9cd":{"kind":"REWORK","_number":2,"created":"2020-11-12 17:23:46.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/97/762397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/762397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/762397/2"}}},"commit":{"parents":[{"commit":"32ee3f1154a7fb1278e60efe544f8e94132a6d65","subject":"[masakari] Release stable deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32ee3f1154a7fb1278e60efe544f8e94132a6d65"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:27:46.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-12 17:23:30.000000000","tz":60},"subject":"[masakari] Transition Stein to EM","message":"[masakari] Transition Stein to EM\n\nThis transitions the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I732722b866234ca5d034272a18ca787a1106986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95b49dfc56ab156212927bb5780faee1e743a9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95b49dfc56ab156212927bb5780faee1e743a9cd"}]},"branch":"refs/heads/master"},"14c54fe643839d1149e7df669b22566c4b903b8b":{"kind":"REWORK","_number":3,"created":"2020-11-13 12:47:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/97/762397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/762397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/762397/3"}}},"commit":{"parents":[{"commit":"535f8c865e7b7fa4739398621a8b77ac3fb65413","subject":"Re-release Masakari Monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/535f8c865e7b7fa4739398621a8b77ac3fb65413"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:27:46.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-13 12:46:59.000000000","tz":60},"subject":"[masakari] Transition Stein to EM","message":"[masakari] Transition Stein to EM\n\nThis transitions the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I732722b866234ca5d034272a18ca787a1106986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14c54fe643839d1149e7df669b22566c4b903b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14c54fe643839d1149e7df669b22566c4b903b8b"}]},"branch":"refs/heads/master"},"51b9905be9b82f0d1be9712f7106403c7d928fae":{"kind":"REWORK","_number":4,"created":"2020-11-17 15:28:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/97/762397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/762397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/762397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/762397/4"}}},"commit":{"parents":[{"commit":"2d39b59555422660024e55303c80eef69f83abb7","subject":"[masakari] Add Victoria cycle highlights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d39b59555422660024e55303c80eef69f83abb7"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 19:27:46.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-17 15:27:48.000000000","tz":60},"subject":"[masakari] Transition Stein to EM","message":"[masakari] Transition Stein to EM\n\nThis transitions the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I732722b866234ca5d034272a18ca787a1106986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51b9905be9b82f0d1be9712f7106403c7d928fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51b9905be9b82f0d1be9712f7106403c7d928fae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
