)]}'
{"id":"openstack%2Freleases~762407","triplet_id":"openstack%2Freleases~master~I840bacd1f0720a670afd072ef7e6a1c311494bb1","project":"openstack/releases","branch":"master","topic":"stein-em","hashtags":[],"change_id":"I840bacd1f0720a670afd072ef7e6a1c311494bb1","subject":"[octavia] Transition Stein to EM","status":"MERGED","created":"2020-11-11 20:04:46.000000000","updated":"2020-11-19 11:22:25.000000000","submitted":"2020-11-19 11:22:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762407-1605784945500-795c03e7","meta_rev_id":"11b18911862cdb4871d92520ceb04356453d60ba","_number":762407,"virtual_id_number":762407,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 11:22:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-18 17:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-11-12 07:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-11-19 10:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-11-12 01:09:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-11-19 10:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-11-12 01:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 20:20:07.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-11-12 01:09:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-12 07:33:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-18 17:12:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-19 10:01:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-19 11:22:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fad1510db499ed7eb5b87250cb2ea8e41bc695d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-11 20:04:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8536616725326eb54f351de79f7473071d9249ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 20:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5751e4a5f234f0e87e573993a45dda7 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55e387cb287547f6a431b48a3fa899e4 : FAILURE in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/68a28a1c599542b8879be3b08d6d249b : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"299f1d12056d0f8ef40ede3eaada895889298953","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-11-11 22:33:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b1b3f4ecf5eb5f5fcf5689411a21c9d20edf744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 22:50:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548758ef88dc40a9a019363c68da80c3 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9304346bbe194756b74e6357d5534662 : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9641e4263d2040c6b92c17800d581203 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"db8715a156fdd3ec98e7110d8675d5e76eaaf00e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-12 01:09:36.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"e4235666392a1eb40d8e67a98cebda40ae83e1b8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 01:10:35.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"5d341721299c568f4a66288d89d23927c7ee5775","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-12 07:33:48.000000000","message":"Patch Set 2: Code-Review+1\n\nNo pending backport patches open. OK from my side.","accounts_in_message":[],"_revision_number":2},{"id":"efcfcf3847b44055ae6642a29c9a346467ab9d40","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-12 08:43:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9df12e571c994df6cd2efe6feb16f50ec1c379d3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-12 14:53:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nHello,\n\nIn my previous review I missed that some patches remains opened, -1 to highlight them before moving to EM","accounts_in_message":[],"_revision_number":2},{"id":"633bcef00dc25d4e8ed256f708ed529cbeccb584","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-18 17:12:57.000000000","message":"Patch Set 2: Code-Review+2\n\nSome decent things that could be released, but they can still be merged and just not included in an official release. That way if someone needs them, they can just pick up those patches.\n\nSince we are past the published transition date for stein EM, unless the team wants to quick get those merged and released, I think we should go ahead with tagging.","accounts_in_message":[],"_revision_number":2},{"id":"50cc7a4742de5d57aa6f913d8b77f8bd1fe0eb0b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 10:01:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOk let\u0027s go","accounts_in_message":[],"_revision_number":2},{"id":"364abbde85d47d051b52d81aaaaf5f3a3c5d71b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 10:02:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ceb4231ba93971b40d6b80dae061d828d54d56a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 10:40:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2cfc7f673cc452b8526047947ab83e8 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6889b4e4bbfe4dbd9d777e71bc19a45a : FAILURE in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51af192645ad444caf45562cb03ffbc0 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c0bd6eb05dcfce56c253c668c68de6529a73eaf3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 10:53:10.000000000","message":"Patch Set 2:\n\nrecheck\nNot sure that doc issue is legit, let\u0027s recheck once.","accounts_in_message":[],"_revision_number":2},{"id":"96ac93c415b5dd1d09624204eaa196a852d1c73d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:10:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1944e0f74a44fbca8633c93311e909b : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92009197f50843a7b286f4edd4f05943 : SUCCESS in 4m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f84449c812474ef39766557413affca2 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f53b9526d28d6f428c486ddf45bd84943a7555d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:10:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81e812829268562228f117f43421f3a87b9c819e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:22:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285d13a1a4f042c0ac662dae0472e978 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6235eb217f0b4143a58dc58efb750897 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47a2fb70cb30438c9a78ea6b41a765ec : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b1086a3b5dcebb24ed6b5fbe1f5c455b8ceeaaf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:22:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9472ad359df325d0f8f145656bd17fd99c402cca","revisions":{"7925ae641f01196eb560f28c4fdcfa241ea6a2e0":{"kind":"REWORK","_number":1,"created":"2020-11-11 20:04:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/762407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/762407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/762407/1"}}},"commit":{"parents":[{"commit":"32499bc048a68f4ddc08a8af3d4b0d15d365413d","subject":"Merge \"Final Cinder team releases for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32499bc048a68f4ddc08a8af3d4b0d15d365413d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 20:04:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 20:04:12.000000000","tz":60},"subject":"[octavia] Transition Stein to EM","message":"[octavia] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I840bacd1f0720a670afd072ef7e6a1c311494bb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7925ae641f01196eb560f28c4fdcfa241ea6a2e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7925ae641f01196eb560f28c4fdcfa241ea6a2e0"}]},"branch":"refs/heads/master"},"9472ad359df325d0f8f145656bd17fd99c402cca":{"kind":"REWORK","_number":2,"created":"2020-11-11 22:33:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/762407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/762407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/762407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/762407/2"}}},"commit":{"parents":[{"commit":"32499bc048a68f4ddc08a8af3d4b0d15d365413d","subject":"Merge \"Final Cinder team releases for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32499bc048a68f4ddc08a8af3d4b0d15d365413d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 20:04:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-11-11 22:33:16.000000000","tz":60},"subject":"[octavia] Transition Stein to EM","message":"[octavia] Transition Stein to EM\n\nThis transition the stein branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I840bacd1f0720a670afd072ef7e6a1c311494bb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9472ad359df325d0f8f145656bd17fd99c402cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9472ad359df325d0f8f145656bd17fd99c402cca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
