)]}'
{"id":"openstack%2Freleases~764434","triplet_id":"openstack%2Freleases~master~Ife77612c0ee674dee3ecf7b7b508f49d60adcdb4","project":"openstack/releases","branch":"master","topic":"oslo-transition-independent","hashtags":[],"change_id":"Ife77612c0ee674dee3ecf7b7b508f49d60adcdb4","subject":"Move oslo.i18n repos from wallaby to independent","status":"ABANDONED","created":"2020-11-27 10:09:16.000000000","updated":"2020-11-30 13:06:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"beb6b56b81029c503fdcc5f7e7fe0908e6c2ae57","_number":764434,"virtual_id_number":764434,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-27 10:20:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":-1,"date":"2020-11-30 10:08:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 10:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-27 10:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 10:08:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c8bfb2b408d689abf15a9c263e2dc2a54db91c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-27 10:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb9122a4acd1aecc0ae3b5f4424cb010cf8180cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 10:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccdc61ce7a045569a2a111fdbbcf227 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f29634d9b26845aba423cd31ebca0aa5 : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f65b374b0310439d89ec4fdc3cf29b39 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"60e13db49353e5739a8054a9fdcecb53e9242bc9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-27 15:24:34.000000000","message":"Patch Set 1:\n\nI want to use this patch as a draft to design the template that I\u0027ll use for the other transitions [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018745.html","accounts_in_message":[],"_revision_number":1},{"id":"2f24b044c6175baaf98d317b619d8bfcdf466772","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-30 10:08:42.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we should not remove the past releases from the coordinated release files. I know this has been done in the past, but IIRC that were cases where the library should probably never have been released as part of the coordinated release (pbr), not mindful transitions like here.\n\nSo I\u0027d recommend just creating am empty _independent file and removing the wallaby placeholder file.","accounts_in_message":[],"_revision_number":1},{"id":"beb6b56b81029c503fdcc5f7e7fe0908e6c2ae57","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-30 13:06:12.000000000","message":"Abandoned\n\nOutdated patch, the transition have been managed with https://review.opendev.org/c/openstack/releases/+/764624","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f500306aca80e27fd6ad83b19eff01ecce4a07c","revisions":{"5f500306aca80e27fd6ad83b19eff01ecce4a07c":{"kind":"REWORK","_number":1,"created":"2020-11-27 10:09:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/34/764434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/764434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/764434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/764434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/764434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/764434/1"}}},"commit":{"parents":[{"commit":"5e402a60e6716644e7a062ee0da67907dade0a47","subject":"Merge \"Set Stein status to Extended Maintenance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e402a60e6716644e7a062ee0da67907dade0a47"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-27 10:06:22.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-27 10:06:22.000000000","tz":60},"subject":"Move oslo.i18n repos from wallaby to independent","message":"Move oslo.i18n repos from wallaby to independent\n\nRemove victoria|ussuri|train branches and move oslo.i18n\nto indenpendent in order to make them branchless.\n\nChange-Id: Ife77612c0ee674dee3ecf7b7b508f49d60adcdb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f500306aca80e27fd6ad83b19eff01ecce4a07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f500306aca80e27fd6ad83b19eff01ecce4a07c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
