)]}'
{"id":"openstack%2Freleases~764710","triplet_id":"openstack%2Freleases~master~I85aee108623bbc747c46ab087d1def892999ac7d","project":"openstack/releases","branch":"master","topic":"w1-c-w-i","hashtags":[],"change_id":"I85aee108623bbc747c46ab087d1def892999ac7d","subject":"Release oslo.limit for wallaby-1 milestone","status":"ABANDONED","created":"2020-11-30 13:28:43.000000000","updated":"2020-12-03 16:31:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bef19412292fa69d0dd78d9eb5d1331510d1b34e","_number":764710,"virtual_id_number":764710,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-30 13:52:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 13:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 13:52:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6a98607e9627f9bdb49e2cd6d784444fa5918e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-30 13:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e1b0ae0ea18ff91a0144360dddd1dc03ec4d6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 13:52:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4114d09efc407094b3003984fbaf1b : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55b085fd450646f6a7f0ad2e23b881e7 : SUCCESS in 4m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/512a9f8fd8cb4706b7118cdbc73e9c34 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bef19412292fa69d0dd78d9eb5d1331510d1b34e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-03 16:31:26.000000000","message":"Abandoned\n\nNo functional changes will be landed I abandon this patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79047d692e16b194466b93b3e49f4eff592c20af","revisions":{"79047d692e16b194466b93b3e49f4eff592c20af":{"kind":"REWORK","_number":1,"created":"2020-11-30 13:28:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/764710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/764710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/764710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/764710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/764710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/764710/1"}}},"commit":{"parents":[{"commit":"d8998a80320b21178f232e1ce946007208d91a9f","subject":"Merge \"New bugfix release of tooz for ussuri.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d8998a80320b21178f232e1ce946007208d91a9f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-30 13:28:31.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-30 13:28:31.000000000","tz":60},"subject":"Release oslo.limit for wallaby-1 milestone","message":"Release oslo.limit for wallaby-1 milestone\n\nThis is a library release for oslo.limit for the wallaby-1\nmilestone. This repo includes commits that have not been releases for\nthis cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\nChange-Id: I85aee108623bbc747c46ab087d1def892999ac7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79047d692e16b194466b93b3e49f4eff592c20af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79047d692e16b194466b93b3e49f4eff592c20af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
