)]}'
{"id":"openstack%2Freleases~765075","triplet_id":"openstack%2Freleases~master~I28f8a4d283ea2c3420ac318eefcce20c9f85339c","project":"openstack/releases","branch":"master","topic":"765075","hashtags":[],"change_id":"I28f8a4d283ea2c3420ac318eefcce20c9f85339c","subject":"Cinder team releases for stable/victoria","status":"MERGED","created":"2020-12-02 10:00:22.000000000","updated":"2020-12-07 09:49:45.000000000","submitted":"2020-12-07 09:49:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765075-765075","meta_rev_id":"6d9b99c120715dba9d054231cb604d52bbabe1e6","_number":765075,"virtual_id_number":765075,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-07 09:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-12-06 16:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-06 16:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-12-07 09:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-12-07 09:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-12-06 16:54:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 10:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 10:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 12:20:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-02 12:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-07 09:27:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"7a40ffc7e3cb8b40bf44e6cd51f08f4682689a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-02 10:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0de2e4325ec75be7014f4c808a98cba455acce00","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-02 10:03:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6ccf4b6bf7aa6a8c89bce98c631a523177dcee41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 10:25:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d229d23c405c46c083cc157b0c68957b : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4474419e8d8493f9be3eb553f062bd7 : SUCCESS in 18m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d1348d51aa0444387d0c731cf5c968c : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8baf796830b39bfec5e85b132a05da7d393703c4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-02 15:02:40.000000000","message":"Patch Set 1:\n\nLet\u0027s hold the release for https://review.opendev.org/c/openstack/cinder/+/764503","accounts_in_message":[],"_revision_number":1},{"id":"999e37b6e35f9035ee109029f267f96efe190d84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-03 15:23:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e31d3448fc6004e76dea13feccb74babb34a139","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-03 15:25:43.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Let\u0027s hold the release for https://review.opendev.org/c/openstack/cinder/+/764503\n\nMerged in stable/victoria, updated the cinder hash","accounts_in_message":[],"_revision_number":2},{"id":"cff0662546e7e372b6e1c70493bcf3dedba1e5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:41:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c972832772a4afb9957b2a68e734473 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5348b4399c974e499e4f6b51770f5f6f : SUCCESS in 13m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/63c7d94ce4514c6abfb1f4e91cca77f1 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"43b9a088e0cc829a0d01539b08c4ebfb14f91742","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-04 15:32:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f6e1d561da8a95d075ad652c2675cd8e5c4ab88c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-04 15:55:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"55093de79f22b585ba7018b9c6194d1343a53e72","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-04 15:56:49.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nMy bad, I included changes that weren\u0027t needed for the release. Updated deliverables now.","accounts_in_message":[],"_revision_number":3},{"id":"547b842ca8500b78d7b36aba526f9c0c0870406c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:29:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9588f90c10464cb55c75067fcf9516 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/449dc4ec305543b8830c8fa5292eb57b : SUCCESS in 14m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/640a36217e0941ab8c830520bcbc31c4 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"20ddb3fc09ac779e546e4728efece3b3c81766f1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-06 16:53:51.000000000","message":"Patch Set 3: Code-Review+1\n\nVerified hash, LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"1ad452fdd3b47995f9c38973a3d90d2111a9fb52","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 16:54:49.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"ae72478bec05b3a52d9cfd47640fa48fa24b6218","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-06 16:55:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b5050d6ad18600a419c7c73d9d21c7011a08380b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-07 09:27:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7eb6b1fe180557822b2217d427952bc2eb3edddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:28:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d35dad1e08acb64a641c96818f14421736960262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:49:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47600e93ac72479092c996248ca6f8d3 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f671016db61461f8292dbff66ae5f55 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec2e92f23c2b43fe8200862722341a4c : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6d9b99c120715dba9d054231cb604d52bbabe1e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 09:49:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"afbac09f56c54c1dd7d6ed776fb6eb52cb9e0141","revisions":{"c0d37b3a670fdadf18622f01e79d48b607187c33":{"kind":"REWORK","_number":1,"created":"2020-12-02 10:00:22.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/75/765075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/765075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/765075/1"}}},"commit":{"parents":[{"commit":"d6d823908fb20694148db5d50e57757e204e8dce","subject":"Merge \"Release python-freezerclient for wallaby-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6d823908fb20694148db5d50e57757e204e8dce"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-02 09:57:21.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-02 10:00:15.000000000","tz":-300},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (cycle trailing - will have a separate release patch)\n- os-brick (no functional changes in branch)\n\nChange-Id: I28f8a4d283ea2c3420ac318eefcce20c9f85339c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0d37b3a670fdadf18622f01e79d48b607187c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0d37b3a670fdadf18622f01e79d48b607187c33"}]},"branch":"refs/heads/master"},"85bfdb7561a2e54c53eb025030d24334c05b0d04":{"kind":"REWORK","_number":2,"created":"2020-12-03 15:23:24.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/75/765075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/765075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/765075/2"}}},"commit":{"parents":[{"commit":"d6d823908fb20694148db5d50e57757e204e8dce","subject":"Merge \"Release python-freezerclient for wallaby-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6d823908fb20694148db5d50e57757e204e8dce"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-02 09:57:21.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-03 15:23:10.000000000","tz":-300},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (cycle trailing - will have a separate release patch)\n- os-brick (no functional changes in branch)\n\nChange-Id: I28f8a4d283ea2c3420ac318eefcce20c9f85339c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/85bfdb7561a2e54c53eb025030d24334c05b0d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/85bfdb7561a2e54c53eb025030d24334c05b0d04"}]},"branch":"refs/heads/master"},"afbac09f56c54c1dd7d6ed776fb6eb52cb9e0141":{"kind":"REWORK","_number":3,"created":"2020-12-04 15:55:45.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/75/765075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/765075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/765075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/765075/3"}}},"commit":{"parents":[{"commit":"d6d823908fb20694148db5d50e57757e204e8dce","subject":"Merge \"Release python-freezerclient for wallaby-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6d823908fb20694148db5d50e57757e204e8dce"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-02 09:57:21.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-04 15:55:41.000000000","tz":-300},"subject":"Cinder team releases for stable/victoria","message":"Cinder team releases for stable/victoria\n\nNot being released at this time:\n- cinderlib (cycle trailing - will have a separate release patch)\n- os-brick (no functional changes in branch)\n- python-cinderclient (no functional changes in branch)\n- python-brick-cinderclient-ext (no functional changes in branch)\n\nChange-Id: I28f8a4d283ea2c3420ac318eefcce20c9f85339c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/afbac09f56c54c1dd7d6ed776fb6eb52cb9e0141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/afbac09f56c54c1dd7d6ed776fb6eb52cb9e0141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
