)]}'
{"id":"openstack%2Freleases~768538","triplet_id":"openstack%2Freleases~master~I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62","project":"openstack/releases","branch":"master","topic":"tempest-plugin-stein-last","hashtags":[],"change_id":"I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62","subject":"[Tempest Plugins] Tag stein-last","status":"MERGED","created":"2020-12-26 18:32:14.000000000","updated":"2021-01-21 09:05:33.000000000","submitted":"2021-01-21 09:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"768538-tempest-plugin-stein-last","meta_rev_id":"f8e306918bc12909e1230a199fe2bd9155425183","_number":768538,"virtual_id_number":768538,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 09:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-01-21 08:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-01-13 08:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-01-21 08:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-01-21 09:05:33.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 18:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-26 18:43:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-26 19:22:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-12-29 10:41:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-21 08:54:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"24e6b76793ca7200757225cf9f5d8a04e3ed7649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 18:32:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3da8020cf152d49256b3ee399b7a31a4f75015b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 18:43:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6a69c451624e529e19ab2c98cb45d0 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc6cd081167149d2a621fec285eb5940 : FAILURE in 3m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d92bfd85553a437698e782c47c47288c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"073b483321027d4820a68354dd8beb8da7862b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:22:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a24bccc0d38d5a4f5f47146a2fd489e01bc78173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 20:36:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367b95b1afe24a5bb549d596fc42d626 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f54ec3f4768a4a0a992303d03ec3d28f : SUCCESS in 3m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f93e185401774a72ab43323c26681111 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6913ef76d2b537c3dd7812f0994cdfc406b4a7df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-28 21:31:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ba2078ad60427d0d5fc4f23d9e6017f10025297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 21:41:41.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6b1509e78347a396a9477055e30e4f : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/617b32847c074802a5b2b27abea6cdc0 : FAILURE in 3m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/654d45dc5cfe4f6f8a3b7965b7c6b5a5 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"45b0929b28ecb3bd21e10b0291b924c8e12d9f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-28 22:01:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c7a3f790ccb2d6a4c9638e60958fb448ff10d107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 22:07:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acddef7a0384922a2a7fd41b0424a55 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d4a10fc5bb64071a6e452edccfa0f84 : SUCCESS in 3m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/470a04a02dbb442e90dc3c8bc4da9c58 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"1c77b8b5e3f8684f45337d2a502d7089e46cd8cb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-29 10:41:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"37d67fabaafd111be94418221f444edc1d7d2a46","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 15:25:31.000000000","message":"Patch Set 4: Code-Review-1\n\nThe blazar-tempest-plugin-stein job is currently failing, so I would like to hold off on this to make sure the stein-em tag is functional.","accounts_in_message":[],"_revision_number":4},{"id":"2fe54e65bfc6ed663d17cd5056e4573260b40983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 22:16:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5c4afbfa9ad27c442b291667c9154e2d29d6babc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 22:16:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"98d71cb40d36a2bb7683593bd2b5a010b468dc77","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 22:18:13.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"f1c1fb47422ac8562ab7b751a1b864354a7ec089","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 22:22:43.000000000","message":"Patch Set 5: Code-Review-1\n\nI suppose we should merge the doc/requirements.txt change as well.","accounts_in_message":[],"_revision_number":5},{"id":"f636dc5cb903f194a8b0d512236849435c3472a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 22:39:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d8ebb39b114142a6607f099c90a6db : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c16da829531e471e91b086d572e4f0fe : FAILURE in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c0e84020a35d468984c94afda5ea2ca4 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c90d13bdab876aee0eed0b9a07340797d2c5a84e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-06 10:40:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a4a7059aed006d31c752dd942679b132e43c597","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:48:36.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"450562c8e571a676ca7837b09a42ce64b4d4d495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:57:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a87c42a3e5c456db5fdc6561c6a5e84 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f9f1c29e01b40f384558aa06e542ef7 : FAILURE in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/68f12528b86a4c84a208c2bf6a3a53df : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"7100148cf632e90071dceaffe2d6ea5228076feb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-06 11:17:22.000000000","message":"Patch Set 6:\n\n2 errors found\ndeliverables/stein/blazar-tempest-plugin.yaml: validate_series_em: only branched deliverables can be tagged EM\ndeliverables/stein/blazar-tempest-plugin.yaml: validate_series_em: EM tag must match the last release, tagging d952210ecd83d78a108baa596a256c429c521503, last release c91065e36281490658b73d9aa1f202d3d6804326\n\nDo we need to release 0.6.0 first as a separate patch?","accounts_in_message":[],"_revision_number":6},{"id":"2284d6115d1b873c5d06c9afe1e3840a4fae854c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-06 11:29:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d3632b9d42ebf172ce6193e9c5f66a3a8dfd55a5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 11:30:30.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"c06a180a1dd2910ec92645d719ea60c13b96139b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 11:40:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285a2fb50b6945419e1381f4b8ad977a : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c674aac2e334d749f144e069e0d90ac : FAILURE in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6957662a275848609a80529ede77c0bd : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7},{"id":"bcaa4673d8657f6efa927d271f187ff5b1dbdf8a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-06 11:48:58.000000000","message":"Patch Set 7:\n\nI assume this will pass validation when https://review.opendev.org/c/openstack/releases/+/769524/ is merged.","accounts_in_message":[],"_revision_number":7},{"id":"847d1c02638d6ba3fb3602be05f03dd0b3cd2ec7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-12 07:37:30.000000000","message":"Patch Set 7:\n\n(1 comment)\n\n\u003e Patch Set 7:\n\u003e \n\u003e I assume this will pass validation when https://review.opendev.org/c/openstack/releases/+/769524/ is merged.\n\nUnfortunatelly we need more changes to fix the validation, c.f my inline comment.","accounts_in_message":[],"_revision_number":7},{"id":"6a870bacd189a6a6ff012eafd9d0fcd78914be5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 08:40:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c3d25aeb2052b44f226881f5101390324b097458","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 08:41:46.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"ce9d869598c75680f0385d61df57cdd3b5b75733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 08:52:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc106793f254c2db59a593ca577c881 : FAILURE in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/635d37a483484dad9682e7c1ec90bba4 : FAILURE in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b79361019ffb450f8ef0f0bddba3e735 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":8},{"id":"407661a1911265249578c6c2655b0fe7eb9b584a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-12 10:55:29.000000000","message":"Patch Set 8:\n\nWell, I modified our machinery to allow us to tag branches as \"-last\" (c.f https://review.opendev.org/c/openstack/project-team-guide/+/769821).\n\nI tested my changes locally against this patch and it works as expected, however my patch depends-on https://review.opendev.org/c/openstack/project-team-guide/+/769821 .\n\nI think we can rebase these changes on https://review.opendev.org/c/openstack/releases/+/770265 , normally it will make the CI happy","accounts_in_message":[],"_revision_number":8},{"id":"7fe0ccaa653c43861ed90922ab7e66e33bdbdb12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 10:57:00.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"db24cb1c2b3a3497db36632d1d4aba117dc7039b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:09:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc48af6d21b4c079353e9ec2bafdbd8 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc057902c2824ae29b84f56ae61ce3d1 : SUCCESS in 6m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/632b192ccb7345b5b9407417c807099a : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":9},{"id":"1a317b7bc72d3da5fedc9fc086e4c242965cf87e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 15:55:34.000000000","message":"Patch Set 9:\n\nalso we need to release the tag with version number so that it can be used in both way, name or particular number","accounts_in_message":[],"_revision_number":9},{"id":"986cac2faef8ee5e3da1a3201e4d99f18ee79894","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 15:57:53.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e also we need to release the tag with version number so that it can be used in both way, name or particular number\n\nah you already released that with same hash, we are good here - https://review.opendev.org/c/openstack/releases/+/769524","accounts_in_message":[],"_revision_number":9},{"id":"e1ae143317c66ca3e234232a49fb4335834d1993","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:20:24.000000000","message":"Topic tempest-plugin-stein-em removed","accounts_in_message":[],"_revision_number":9},{"id":"3097ff241bad5734bcb218e735fec4621f1f601b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:20:26.000000000","message":"Topic set to tempest-plugin-stein-last","accounts_in_message":[],"_revision_number":9},{"id":"ac5070aa47c849288ffab20179872d7dfb11c729","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-13 08:48:46.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"444ad6465c859560e9f7c330c3bb8bacfd16d06c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-21 08:54:17.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"781a57fc340484dc427e735df0a6b89ad1299bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 08:54:51.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8f5dfe33535f0208b88dc7ae9269480bff3f185b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:05:26.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f4320bef994bc4bafb152025f804ed : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5277e4b83c704ca9bedac768e7fcfa5b : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3b1bdc7cdc148958c64ed2439883189 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":9},{"id":"9ad119fc3a09a84c6621f1144e709a51473baf84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:05:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"437bb5a7d2c16ebc65167c0b80535f9a9c8ddbd7","revisions":{"079147c80e91a7cb66ad08e2bb694c7293305644":{"kind":"REWORK","_number":1,"created":"2020-12-26 18:32:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/768538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/1"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/079147c80e91a7cb66ad08e2bb694c7293305644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/079147c80e91a7cb66ad08e2bb694c7293305644"}]},"branch":"refs/heads/master"},"a592b7e3a0872c9f2368ba370b46371d7a8ea0e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-26 20:22:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/768538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/2"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-26 20:22:32.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a592b7e3a0872c9f2368ba370b46371d7a8ea0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a592b7e3a0872c9f2368ba370b46371d7a8ea0e1"}]},"branch":"refs/heads/master","description":"Rebase"},"9a8b4b6af0e760fda9651e263c67dc651cf395c4":{"kind":"REWORK","_number":3,"created":"2020-12-28 21:31:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/768538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/3"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-28 21:31:27.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a8b4b6af0e760fda9651e263c67dc651cf395c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a8b4b6af0e760fda9651e263c67dc651cf395c4"}]},"branch":"refs/heads/master"},"062fc469441cd867d75d3259ccfeb397bf29c6fc":{"kind":"REWORK","_number":4,"created":"2020-12-28 22:01:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/768538/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/4"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-28 22:01:27.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/062fc469441cd867d75d3259ccfeb397bf29c6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/062fc469441cd867d75d3259ccfeb397bf29c6fc"}]},"branch":"refs/heads/master"},"b0f934444f56122b9582be1c9d4144ff42cfc42c":{"kind":"REWORK","_number":5,"created":"2021-01-05 22:16:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/768538/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/5"}}},"commit":{"parents":[{"commit":"00834b81248a1f60130f64190026c2835d224064","subject":"Merge \"TripleO repos feature release for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00834b81248a1f60130f64190026c2835d224064"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-05 22:16:30.000000000","tz":60},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f934444f56122b9582be1c9d4144ff42cfc42c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f934444f56122b9582be1c9d4144ff42cfc42c"}]},"branch":"refs/heads/master"},"f57eff2a678fc9463c40feb5fd34176b0e74b830":{"kind":"REWORK","_number":6,"created":"2021-01-06 10:40:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/768538/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/6"}}},"commit":{"parents":[{"commit":"00834b81248a1f60130f64190026c2835d224064","subject":"Merge \"TripleO repos feature release for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00834b81248a1f60130f64190026c2835d224064"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-06 10:40:35.000000000","tz":60},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f57eff2a678fc9463c40feb5fd34176b0e74b830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f57eff2a678fc9463c40feb5fd34176b0e74b830"}]},"branch":"refs/heads/master"},"b1f3a2c3f43c4d82e1982f4cd4a2205af256d01d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2021-01-06 11:29:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/768538/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/7"}}},"commit":{"parents":[{"commit":"94aa65d5c037f4306b636bdefe003bbbff346ce3","subject":"Release blazar-tempest-plugin for wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94aa65d5c037f4306b636bdefe003bbbff346ce3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-06 11:28:12.000000000","tz":60},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 with the commit used for 0.6.0.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1f3a2c3f43c4d82e1982f4cd4a2205af256d01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1f3a2c3f43c4d82e1982f4cd4a2205af256d01d"}]},"branch":"refs/heads/master"},"64e8fe6a6da5910679b59d05fb1ea7dda44817b4":{"kind":"REWORK","_number":8,"created":"2021-01-12 08:40:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/768538/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/8"}}},"commit":{"parents":[{"commit":"d851b92474a29443386bfb740e78cfd9bf7d20d7","subject":"Merge \"Release blazar-tempest-plugin for wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d851b92474a29443386bfb740e78cfd9bf7d20d7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 08:39:43.000000000","tz":60},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 with the commit used for 0.6.0.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64e8fe6a6da5910679b59d05fb1ea7dda44817b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/64e8fe6a6da5910679b59d05fb1ea7dda44817b4"}]},"branch":"refs/heads/master"},"437bb5a7d2c16ebc65167c0b80535f9a9c8ddbd7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-12 10:57:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/768538/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/768538/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/768538/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/768538/9"}}},"commit":{"parents":[{"commit":"8d61d2482b153c34419bcd95701e9f262496e441","subject":"Allow to tag projects as \"last\" on EM branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d61d2482b153c34419bcd95701e9f262496e441"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:31:29.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 10:56:50.000000000","tz":60},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 with the commit used for 0.6.0.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I95ef209e4f1a6fabd01eab2d796c5b933f1c3b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/437bb5a7d2c16ebc65167c0b80535f9a9c8ddbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/437bb5a7d2c16ebc65167c0b80535f9a9c8ddbd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
