)]}'
{"id":"openstack%2Freleases~768543","triplet_id":"openstack%2Freleases~master~Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321","project":"openstack/releases","branch":"master","topic":"tempest-plugin-stein-last","hashtags":[],"change_id":"Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321","subject":"[Tempest Plugins] Tag stein-last","status":"MERGED","created":"2020-12-26 18:47:49.000000000","updated":"2021-01-25 10:15:49.000000000","submitted":"2021-01-25 10:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768543-tempest-plugin-stein-last","meta_rev_id":"8a4226afe6c1c03c66a0c85a35352b5faaf8fec4","_number":768543,"virtual_id_number":768543,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 10:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-01-25 10:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-01-13 09:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-23 03:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-01-25 10:04:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-01-23 03:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 18:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-26 18:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-26 19:21:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-12-26 19:22:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2020-12-29 10:58:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-25 10:04:12.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"2156691469788cfea63c417dde839635306bb922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 18:47:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18b15def30a1b7280429a0b44d600f50f58dbf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 18:58:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a52e9a425b7474a844252006cf10881 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e25a388c47784b49bfa8795e11f892fb : FAILURE in 4m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/73789346f34243bc9f9bd6637e3e030e : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"23ec3c18972a1549d6258f087293eaefd77d78c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:23:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fdf4bbe4f6709679a66c90ddc3ba314a56d8e596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 20:32:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9086a9e4884151b2497665cac5efa6 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b8e703c6deb4d0bbdeae0bc87f7f4b3 : SUCCESS in 3m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fb5615230c8d4cfbad75b139bb688989 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"91f3bb412d7925fc891b11149c3a33d39c9db2bb","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-12-27 23:54:26.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"6813edf153b0680b12496f97c32cbb090267741f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-27 23:55:17.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1baee44b5ea9d73934a0bf44fa69e4ca76301db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-28 21:19:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59ff06f5d9fc4d32fa74f4f2835f0d84a4d49752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 21:31:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2682a7d20346d8afb8f84f58282517 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/518c6d11821c4cad847c72e02237df64 : SUCCESS in 4m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d846473552104218809d4f035a3ad808 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e43c2fde3abe0155008255e929cf063303f9ef47","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-29 10:58:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2bcb9a1e0ea3dc86056f94350921a15d25513aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 22:23:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"aae55f15983d59e6f19ed6a2d69609f9a7c5be7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 22:24:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"602ce167c8e57226a645db9ff8dfd0c53190be82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 22:56:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfe0ad0966f4754a232a04ff362d7e2 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04d2102576984ab680483590adcb8e0d : SUCCESS in 6m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4a066b51dcc8495f8fc5cc9eccf5ba03 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":5},{"id":"d4f081bd603cb938fe4b366c5fe5c55b7b714a25","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:18:05.000000000","message":"Topic tempest-plugin-stein-em removed","accounts_in_message":[],"_revision_number":5},{"id":"62cd7500b7e7352690fca87b0f2a40da5ee2230f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:18:14.000000000","message":"Topic set to tempest-plugin-stein-last","accounts_in_message":[],"_revision_number":5},{"id":"8ab245caec3ae91cc6e38859db7002a2eb48c85f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-13 09:29:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5b1d7714c3f8548bd774256d527f155606c1ee2f","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-01-23 03:43:34.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM.\nThanks","accounts_in_message":[],"_revision_number":5},{"id":"73c005258b7f4292159e5fa51b586689044aaff2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 03:44:35.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"aba120d1563b4fe6f958019de1404a6e8fb66394","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-25 10:04:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fa6fa4fed34d6a908a3b2f88973aa0b5fea21345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:04:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3c4058a30ac1c9fec65ad0d7b4b3a5cbb006995f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:15:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1aab0ce0cb14ba3acfca010dc67b000 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412770823dab40ba8e33f07ef989dd4a : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bddb6264e45542b4891b378e3f776464 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8a4226afe6c1c03c66a0c85a35352b5faaf8fec4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"629acf42d0339af3bf04c71819f9a434ec4713e4","revisions":{"4631af81ea34ac3a1ae571e18df537c6195f20b3":{"kind":"REWORK","_number":1,"created":"2020-12-26 18:47:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/768543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/768543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/768543/1"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4631af81ea34ac3a1ae571e18df537c6195f20b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4631af81ea34ac3a1ae571e18df537c6195f20b3"}]},"branch":"refs/heads/master"},"13ca947ab9817946168ea94d860ca109a9e84e5b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-26 20:23:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/768543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/768543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/768543/2"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-26 20:23:12.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13ca947ab9817946168ea94d860ca109a9e84e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13ca947ab9817946168ea94d860ca109a9e84e5b"}]},"branch":"refs/heads/master","description":"Rebase"},"b72e97142524f51fb67076a1733132c85cd7e543":{"kind":"REWORK","_number":3,"created":"2020-12-28 21:19:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/768543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/768543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/768543/3"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-28 21:19:34.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b72e97142524f51fb67076a1733132c85cd7e543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b72e97142524f51fb67076a1733132c85cd7e543"}]},"branch":"refs/heads/master"},"a328b968d3e97517c8ae6c92b9346647dc7967b9":{"kind":"REWORK","_number":4,"created":"2021-01-12 22:23:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/768543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/768543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/768543/4"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-12 22:22:54.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 as well as a new version also with same hash.\nDocumenting this new tag name here\n- https://review.opendev.org/c/openstack/project-team-guide/+/769821\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a328b968d3e97517c8ae6c92b9346647dc7967b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a328b968d3e97517c8ae6c92b9346647dc7967b9"}]},"branch":"refs/heads/master"},"629acf42d0339af3bf04c71819f9a434ec4713e4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-12 22:24:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/768543/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/768543/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/768543/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/768543/5"}}},"commit":{"parents":[{"commit":"8d61d2482b153c34419bcd95701e9f262496e441","subject":"Allow to tag projects as \"last\" on EM branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d61d2482b153c34419bcd95701e9f262496e441"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:47:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-12 22:24:37.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 as well as a new version also with same hash.\nDocumenting this new tag name here\n- https://review.opendev.org/c/openstack/project-team-guide/+/769821\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: Ifaba5f4e460c2a8ee8d00d430678f82dbfb6b321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/629acf42d0339af3bf04c71819f9a434ec4713e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/629acf42d0339af3bf04c71819f9a434ec4713e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
