)]}'
{"id":"openstack%2Freleases~768544","triplet_id":"openstack%2Freleases~master~I3dedf43f254a745684014bc050ff6ac399144b06","project":"openstack/releases","branch":"master","topic":"tempest-plugin-stein-last","hashtags":[],"change_id":"I3dedf43f254a745684014bc050ff6ac399144b06","subject":"[Tempest Plugins] Tag stein-last","status":"MERGED","created":"2020-12-26 18:48:38.000000000","updated":"2021-02-08 13:16:31.000000000","submitted":"2021-02-08 13:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768544-tempest-plugin-stein-last","meta_rev_id":"4527cfd7c8c08ce8226b486f6b016e0e39e465d0","_number":768544,"virtual_id_number":768544,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 13:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 13:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-28 17:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-01-13 09:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 13:05:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 18:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-26 18:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-26 19:21:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-12-29 11:15:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 17:50:21.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-02-08 13:05:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"b11a6745c0006fc78f8950f73609e35b9c9f9cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 18:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"accae0ca5e0401fdd9ccf77d7777a6cf4fbc72a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 18:59:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be03b9cb76ae4507ac77b941b97d8e51 : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a39fd1348e0d4640941a69379aacb9cf : FAILURE in 3m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/31439845c9c842e997694634a2193bdb : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f86ab0f71e89ce3746018487e0042464a53079e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:23:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e36e50cdb2d93604e543c4602ca5836b857759b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 20:32:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3cd373d20654098a69a36b45adee1fd : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9409b8d04f8042dd9af08ffb69721815 : SUCCESS in 4m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c214bc2b38744fdd94c7aeb52cbd26cc : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c1d15ff8782a327113077769b74a4dade743c219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-28 21:50:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d40a9621be94a6f0ff44512254681cd7dd6b6ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 21:57:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1293547f70ac435b8eb90bd4956a1e8c : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/763801ba18c04a1daa8f12039bb3d268 : SUCCESS in 4m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fc2fcb4c7921412e9ec4be021acb0d47 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"587be69aa8822bac6de81303708ae35efb6e7697","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-29 11:15:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1bab603fdac10a94c863e39c9ac09acbeb4f3104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 22:18:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b8ac1c8db976d746003800b001b68d3584d8dc6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 22:18:48.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c36f035f398d1d1ef70e9eda57b2c7d7430bdf38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 22:57:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e6b1ae8109437f814449fc857e4cc9 : SUCCESS in 10m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/270dc4ea3ecc41658dd0be03e9d829f7 : SUCCESS in 4m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ef02d03245d446e393c1ed4af00b99da : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":5},{"id":"2b4c807f4049414b21e9441f22420a82841d221e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:17:45.000000000","message":"Topic tempest-plugin-stein-em removed","accounts_in_message":[],"_revision_number":5},{"id":"8eca2bb338eb2888fce27caa831d5c471894dadb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:17:49.000000000","message":"Topic set to tempest-plugin-stein-last","accounts_in_message":[],"_revision_number":5},{"id":"8b284cd742369c6a3b33d2740acec42eee526429","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-13 09:30:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5693c13e30cc9ebc0ae908f7580489911ad27c79","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-27 12:21:51.000000000","message":"Patch Set 5:\n\nStill waiting for PTL/liaison validation","accounts_in_message":[],"_revision_number":5},{"id":"3966dec3c3ee14ef11a804b94d7bcff6598583f6","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-01-28 17:50:21.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"86deaf5ed2a48ef1a5be4275e31d4e50d2531c10","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-05 13:26:12.000000000","message":"Patch Set 5:\n\nWithout PTL/liaison validation until Monday 8, february we\u0027ll force the merge of these changes","accounts_in_message":[],"_revision_number":5},{"id":"86a3961357293aa8b825eb89714b29dbb45fd208","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 13:05:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nGood to go.","accounts_in_message":[],"_revision_number":5},{"id":"0c326f4c70cd500195c81bad3e12fbab7244aa1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:05:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f977da0c1f6d7de49bc38e6fc8fa7c2dce0f1def","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:16:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ae54a46e6e4022965fb86c980980de : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28cdb5f1ee14397820b8e8676e9bbfa : SUCCESS in 5m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eeff75844cd6441687977cb63b4efac7 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":5},{"id":"4527cfd7c8c08ce8226b486f6b016e0e39e465d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:16:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c35d87c0fef8fd141441d9ef421e0eb3bc4e539","revisions":{"4b424620b9d41a5132aef74ac752496195370a08":{"kind":"REWORK","_number":1,"created":"2020-12-26 18:48:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/768544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/768544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/768544/1"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I3dedf43f254a745684014bc050ff6ac399144b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b424620b9d41a5132aef74ac752496195370a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b424620b9d41a5132aef74ac752496195370a08"}]},"branch":"refs/heads/master"},"1bbdb4488b3d96cbbd2b96bed22597fa968b8182":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-26 20:23:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/768544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/768544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/768544/2"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-26 20:23:25.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I3dedf43f254a745684014bc050ff6ac399144b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bbdb4488b3d96cbbd2b96bed22597fa968b8182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bbdb4488b3d96cbbd2b96bed22597fa968b8182"}]},"branch":"refs/heads/master","description":"Rebase"},"554904a6317303dc31a5e8726aa311cec7918163":{"kind":"REWORK","_number":3,"created":"2020-12-28 21:50:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/768544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/768544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/768544/3"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-28 21:49:54.000000000","tz":-360},"subject":"[Tempest Plugins] Tag stein EM","message":"[Tempest Plugins] Tag stein EM\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-em\u0027 as well as a new version also with same hash.\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I3dedf43f254a745684014bc050ff6ac399144b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/554904a6317303dc31a5e8726aa311cec7918163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/554904a6317303dc31a5e8726aa311cec7918163"}]},"branch":"refs/heads/master"},"9d687ff4cd151671649ad216be64dffb2a3b1bc1":{"kind":"REWORK","_number":4,"created":"2021-01-12 22:18:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/768544/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/768544/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/768544/4"}}},"commit":{"parents":[{"commit":"362db163bd198e298a6282ebf6b857df332e0629","subject":"Allow branchless repo to release EM tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-12 22:14:03.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 as well as a new version also with same hash.\nDocumenting this new tag name here\n- https://review.opendev.org/c/openstack/project-team-guide/+/769821\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I3dedf43f254a745684014bc050ff6ac399144b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d687ff4cd151671649ad216be64dffb2a3b1bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d687ff4cd151671649ad216be64dffb2a3b1bc1"}]},"branch":"refs/heads/master"},"6c35d87c0fef8fd141441d9ef421e0eb3bc4e539":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-12 22:18:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/768544/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/768544/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/768544/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/768544/5"}}},"commit":{"parents":[{"commit":"8d61d2482b153c34419bcd95701e9f262496e441","subject":"Allow to tag projects as \"last\" on EM branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d61d2482b153c34419bcd95701e9f262496e441"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 18:48:23.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-12 22:18:48.000000000","tz":0},"subject":"[Tempest Plugins] Tag stein-last","message":"[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 as well as a new version also with same hash.\nDocumenting this new tag name here\n- https://review.opendev.org/c/openstack/project-team-guide/+/769821\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nChange-Id: I3dedf43f254a745684014bc050ff6ac399144b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c35d87c0fef8fd141441d9ef421e0eb3bc4e539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c35d87c0fef8fd141441d9ef421e0eb3bc4e539"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
