)]}'
{"id":"openstack%2Freleases~768566","triplet_id":"openstack%2Freleases~master~Id575b270e3e3299bf191619bf7e0c9fe8d42a7f4","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Id575b270e3e3299bf191619bf7e0c9fe8d42a7f4","subject":"Allow branchless repo to release EM tag","status":"ABANDONED","created":"2020-12-26 19:26:18.000000000","updated":"2021-01-07 21:15:34.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4f03f61a3cf423e0bc478c8c3aad7c1a7be63d6f","_number":768566,"virtual_id_number":768566,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-30 19:08:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-07 17:19:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2021-01-07 17:48:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 19:34:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-27 10:46:49.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-12-30 12:56:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2020-12-30 12:56:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-04 09:03:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-01-07 17:19:06.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-01-07 17:48:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0e1042057f30795dc9050e9f3cfcc17e60ca8c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 19:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9519a32fbe1d1d3298010e26ce92584e271ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 19:34:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bbe032c369433d8e2fa65d8e8ffcd0 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5806996f2fe644f5aca0bd9eff944e50 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c72e9ae69f243f8858ecc3823ad3761 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849a934ef1994454adb8a09558021222 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9d21cc67ab24230a15e262e8b5d07be : SUCCESS in 3m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/802833990fc143f491708f6b32afc34b : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c70bbb40bf794aea8276d53890c395de : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"87be6289758d5adf6003ff060798a88883f7f90a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:01:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6a72c8d1ed6affd46c7240e6b63e748156e1271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 20:13:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfad71317e54fc8a88e2f8c67173e05 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4944cf708d5426aa7d41cfb475a7ab3 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d4193f5c1740f5a4d3487579a2ac48 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5409b045420475ba7d850118fd89e85 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5707461458b44b19bc8f9e0bfc8f279a : SUCCESS in 3m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aa437a347af949a5b63e5a76cec960e3 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bd4a385df074ac4b3f63f720ba00c2b : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bb4e2b07d5bd91601711aaf31f4151508e4b251c","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-12-27 10:46:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cadd8fdde8ed674abfeb6599b21f3a14b42c9757","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-30 12:56:26.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nWe have never marked these -em before, and I don\u0027t really think we should. There is nothing here to \"extend maintenance\" on. The last real tag is the marker we\u0027ve gone by to know what coincided with which stable release. This is mixing concepts that do not go together.","accounts_in_message":[],"_revision_number":2},{"id":"02154e0801e1bdd8ac6c588a786ddc1a7761eef5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-30 12:56:34.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"454194d35f50cc1ea94045652237e5f4b71fe6c2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-30 19:01:05.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e We have never marked these -em before, and I don\u0027t really think we should. There is nothing here to \"extend maintenance\" on. The last real tag is the marker we\u0027ve gone by to know what coincided with which stable release. This is mixing concepts that do not go together.\n\nThese were -em tagged in rocky[1] but those were with wrong hash[2].\n\nWhy we need -em tag for Tempest and its plugins:\nTempest and its plugins (which are branchless) master version does not support the EM branches[3] so we need to release a tag for those EM to use it for their testing until they move to Unmaintained. If we do not release a EM tag for tempest and plugins then it is not easy to maintained their CI testing. I have faced a lot of issue in past due to this. I have explained the versioning things for Tempest and its plugins here- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1]https://github.com/openstack/releases/blob/08361a137a2b9709ada79893bdcc0433abbc9fcd/deliverables/rocky/neutron-tempest-plugin.yaml#L21\n\n[2] https://review.opendev.org/c/openstack/releases/+/709901/4/deliverables/rocky/neutron-tempest-plugin.yaml\n\n[3] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html","accounts_in_message":[],"_revision_number":2},{"id":"9029e53fbbc9872fad7b50cf617d88b14bcfab84","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-30 19:01:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cc28fae29484e2f8215761e8838bfc4891f1d726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-30 19:01:58.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0d7a7c9736a5cf284dce9595eef011c1f650bd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 19:08:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77f641ab8e046b18d8eb8f8de1b39b5 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4cfdd70a342409597c86769a347fdf3 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5172197355ff4e30953a375398a41f26 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966e2f190bd2485e9de01b921a421846 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7722a82b7c0b4f7c973562cc2470b81b : SUCCESS in 3m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7b175daf31fc4039a17c36a00962aa89 : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b4efe86fac1423d8df2356b4e674fec : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"906eef9531397135f69cb21f3324edbcc285e7e9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-04 09:03:05.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e We have never marked these -em before, and I don\u0027t really think we should. There is nothing here to \"extend maintenance\" on. The last real tag is the marker we\u0027ve gone by to know what coincided with which stable release. This is mixing concepts that do not go together.\n\u003e \n\u003e These were -em tagged in rocky[1] but those were with wrong hash[2].\n\u003e \n\u003e Why we need -em tag for Tempest and its plugins:\n\u003e Tempest and its plugins (which are branchless) master version does not support the EM branches[3] so we need to release a tag for those EM to use it for their testing until they move to Unmaintained. If we do not release a EM tag for tempest and plugins then it is not easy to maintained their CI testing. I have faced a lot of issue in past due to this. I have explained the versioning things for Tempest and its plugins here- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\u003e \n\u003e [1]https://github.com/openstack/releases/blob/08361a137a2b9709ada79893bdcc0433abbc9fcd/deliverables/rocky/neutron-tempest-plugin.yaml#L21\n\u003e \n\u003e [2] https://review.opendev.org/c/openstack/releases/+/709901/4/deliverables/rocky/neutron-tempest-plugin.yaml\n\u003e \n\u003e [3] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\nHello,\n\nThanks for the details.\n\nOn a semantical point of view I agree with Sean, it can look weird to tag branchless project as -em, Sean is right this is mixing concepts, but if it can help to manage testing on stale branches it does not shock me to tag these projects as -em. I missed that point on my previous review on the related patches.\n\nI don\u0027t have all the design\u0027s history of -em and which rules are the rule in this case, but I can see that we also tagged queens and ocata as -em in the past ( https://opendev.org/openstack/tempest/src/tag/queens-em ), however I can\u0027t find rocky, maybe the job failed due to the wrong sha...\n\nAnyway I\u0027m not against to continue with -em on branchless as did on the past, but first lets ask if we have an other option to maintain their CI testing? If the response is no, then I think that we should continue with that. Any idea?\n\nIf we decide to continue with these changes I think that we will face another issue with our tooling/process, especially with EOL. Recently we introduced a script to detect EOL stale branches, and I\u0027m afraid that these changes will be a first step towards the EOL path with branchless projects, and here we will face an issue with our shared process with the infra team (removing of EOL branches on a branchless projects).\n\nBefore moving further with these change I think we need some discussions with all the team and even with the infra team to see the benefits and the risk to follow this path.\n\nThoughts?","accounts_in_message":[],"_revision_number":3},{"id":"a06e789408daaef19f625c4c42685cfb3ca4c552","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-07 17:19:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a990a38dd02ce0aa0bc17d70cff9f6ecd06f171d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-07 17:25:12.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e (1 comment)\n\u003e \u003e \u003e \n\u003e \u003e \u003e We have never marked these -em before, and I don\u0027t really think we should. There is nothing here to \"extend maintenance\" on. The last real tag is the marker we\u0027ve gone by to know what coincided with which stable release. This is mixing concepts that do not go together.\n\u003e \u003e \n\u003e \u003e These were -em tagged in rocky[1] but those were with wrong hash[2].\n\u003e \u003e \n\u003e \u003e Why we need -em tag for Tempest and its plugins:\n\u003e \u003e Tempest and its plugins (which are branchless) master version does not support the EM branches[3] so we need to release a tag for those EM to use it for their testing until they move to Unmaintained. If we do not release a EM tag for tempest and plugins then it is not easy to maintained their CI testing. I have faced a lot of issue in past due to this. I have explained the versioning things for Tempest and its plugins here- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\u003e \u003e \n\u003e \u003e [1]https://github.com/openstack/releases/blob/08361a137a2b9709ada79893bdcc0433abbc9fcd/deliverables/rocky/neutron-tempest-plugin.yaml#L21\n\u003e \u003e \n\u003e \u003e [2] https://review.opendev.org/c/openstack/releases/+/709901/4/deliverables/rocky/neutron-tempest-plugin.yaml\n\u003e \u003e \n\u003e \u003e [3] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n\u003e \n\u003e Hello,\n\u003e \n\u003e Thanks for the details.\n\u003e \n\u003e On a semantical point of view I agree with Sean, it can look weird to tag branchless project as -em, Sean is right this is mixing concepts, but if it can help to manage testing on stale branches it does not shock me to tag these projects as -em. I missed that point on my previous review on the related patches.\n\u003e \n\u003e I don\u0027t have all the design\u0027s history of -em and which rules are the rule in this case, but I can see that we also tagged queens and ocata as -em in the past ( https://opendev.org/openstack/tempest/src/tag/queens-em ), however I can\u0027t find rocky, maybe the job failed due to the wrong sha...\n\u003e \n\u003e Anyway I\u0027m not against to continue with -em on branchless as did on the past, but first lets ask if we have an other option to maintain their CI testing? If the response is no, then I think that we should continue with that. Any idea?\n\nIf no em tag then we need to manually try and find the compatible version at our CI/CD as well as in production testing also. I did the same when rocky testing was broken with master version and I need to find compatible Tempest hash/version and for plugins also.\n\n\n\u003e \n\u003e If we decide to continue with these changes I think that we will face another issue with our tooling/process, especially with EOL. Recently we introduced a script to detect EOL stale branches, and I\u0027m afraid that these changes will be a first step towards the EOL path with branchless projects, and here we will face an issue with our shared process with the infra team (removing of EOL branches on a branchless projects).\n\nThis should not impact the EOL decision, if any of the EM stable release move to EOL then we do not need to release new tag for Tempest or plugins as this em tag can be used as last compatible and tested version. That is how it will be used for production cloud testing too.\n\n\u003e \n\u003e Before moving further with these change I think we need some discussions with all the team and even with the infra team to see the benefits and the risk to follow this path.\n\u003e \n\u003e Thoughts?","accounts_in_message":[],"_revision_number":3},{"id":"29629411766ed331e849b69bd59d818e3c4e636b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-07 17:48:58.000000000","message":"Patch Set 3: Code-Review-1\n\njust as a reminder: this is not needed anymore, as not *-em tag, but rather *-last tag will be used for as tempest support indicator.","accounts_in_message":[],"_revision_number":3},{"id":"af6b5aab0289d9cca3b44e7c99602233f042acf0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-07 18:03:23.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e just as a reminder: this is not needed anymore, as not *-em tag, but rather *-last tag will be used for as tempest support indicator.\n\nyeah, we can abandon this now as we discussed in release meeting to use *-last for tempest and plugins\n- http://eavesdrop.openstack.org/meetings/releaseteam/2021/releaseteam.2021-01-07-17.00.log.html#l-46","accounts_in_message":[],"_revision_number":3},{"id":"4f03f61a3cf423e0bc478c8c3aad7c1a7be63d6f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-07 21:15:34.000000000","message":"Abandoned\n\ndefining new tag for branchless EM https://review.opendev.org/c/openstack/project-team-guide/+/769821","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffe0d892090b13c7ce7f24dda650c565f9fb2188","revisions":{"e1880f13a52381949d7d622fe90d79b6f06d19bf":{"kind":"REWORK","_number":1,"created":"2020-12-26 19:26:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/768566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/768566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/768566/1"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 19:22:59.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 19:22:59.000000000","tz":-360},"subject":"Allow branchless repo to release EM tag","message":"Allow branchless repo to release EM tag\n\nTempest and its plugins are branchless but\nneed to release the EM tag\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019565.html\n\nThis commit allow to tag EM for branchless repo too.\n\nChange-Id: Id575b270e3e3299bf191619bf7e0c9fe8d42a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1880f13a52381949d7d622fe90d79b6f06d19bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1880f13a52381949d7d622fe90d79b6f06d19bf"}]},"branch":"refs/heads/master"},"362db163bd198e298a6282ebf6b857df332e0629":{"kind":"REWORK","_number":2,"created":"2020-12-26 20:01:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/768566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/768566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/768566/2"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 19:22:59.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 19:59:44.000000000","tz":-360},"subject":"Allow branchless repo to release EM tag","message":"Allow branchless repo to release EM tag\n\nTempest and its plugins are branchless but\nneed to release the EM tag\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019565.html\n\nThis commit allow to tag EM for branchless repo with\nlatest hash.\n\nChange-Id: Id575b270e3e3299bf191619bf7e0c9fe8d42a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/362db163bd198e298a6282ebf6b857df332e0629"}]},"branch":"refs/heads/master"},"ffe0d892090b13c7ce7f24dda650c565f9fb2188":{"kind":"REWORK","_number":3,"created":"2020-12-30 19:01:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/768566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/768566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/768566/3"}}},"commit":{"parents":[{"commit":"3b07f219398879cfb910c0691009f720acfaf2e2","subject":"Create RC1 for OpenStack-Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b07f219398879cfb910c0691009f720acfaf2e2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-26 19:22:59.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-30 19:01:55.000000000","tz":0},"subject":"Allow branchless repo to release EM tag","message":"Allow branchless repo to release EM tag\n\nTempest and its plugins are branchless but\nneed to release the EM tag\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019565.html\n\nThis commit allow to tag EM for branchless repo with\nlatest hash.\n\nChange-Id: Id575b270e3e3299bf191619bf7e0c9fe8d42a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffe0d892090b13c7ce7f24dda650c565f9fb2188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffe0d892090b13c7ce7f24dda650c565f9fb2188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
