)]}'
{"id":"openstack%2Freleases~768766","triplet_id":"openstack%2Freleases~master~I4c214340aab240427b8e3fa87fd91d8e688d92cf","project":"openstack/releases","branch":"master","topic":"765096","hashtags":[],"change_id":"I4c214340aab240427b8e3fa87fd91d8e688d92cf","subject":"Release cinderlib 3.0.0","status":"MERGED","created":"2020-12-30 04:16:28.000000000","updated":"2021-01-14 13:40:10.000000000","submitted":"2021-01-14 13:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768766-765096","meta_rev_id":"1d3ee4772299ca44885fb9568e29989921d3fe74","_number":768766,"virtual_id_number":768766,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 13:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-14 13:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-14 13:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-01-14 09:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-14 09:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-01-14 13:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-01-14 13:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-30 04:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-30 04:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-30 12:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-30 12:20:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-04 08:18:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-14 09:21:31.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"f7fe46f38e28339aea6a7d788835b258b792bb54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-30 04:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"297fc40a1194c7abf6c6864d9d56d9572645e103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 04:24:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408e5681e76b47cb972d2a5856fbd723 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/121a9f59cc954fe883984ae02b6acfcc : SUCCESS in 4m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e01ea206a7a2461097d08e6239e517b2 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f12cdcdd461fbbfa4005e41b5cd781f356900009","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-04 08:18:20.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e0ddaf2167190ac3d3b4961a12c20f86c7638472","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-04 14:56:43.000000000","message":"Patch Set 1: Code-Review-1\n\nEven though cinderlib doesn\u0027t have a l-c job, I wonder whether we should merge https://review.opendev.org/c/openstack/cinderlib/+/768533 first to get the requirements updated before release.","accounts_in_message":[],"_revision_number":1},{"id":"42d696add8ba9ecc293cf6b2acd37dc8857c0802","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 09:56:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Even though cinderlib doesn\u0027t have a l-c job, I wonder whether we should merge https://review.opendev.org/c/openstack/cinderlib/+/768533 first to get the requirements updated before release.\n\nYes please merge it first it could avoid to fail during our release jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0d810e9e90aa54ad9ec21b2b8aa08da45810b85","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-14 03:31:53.000000000","message":"Patch Set 1:\n\nUpdate: https://review.opendev.org/c/openstack/cinderlib/+/768533 has been revised but has been stuck in the Zuul check all day.  Once it passes and has merged, either Rajat or I will update the hash here.","accounts_in_message":[],"_revision_number":1},{"id":"c0013f0de632a3ae4d13b0d7c5529bc132409220","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-14 06:45:40.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1:\n\u003e \n\u003e Update: https://review.opendev.org/c/openstack/cinderlib/+/768533 has been revised but has been stuck in the Zuul check all day.  Once it passes and has merged, either Rajat or I will update the hash here.\n\nAck, thanks","accounts_in_message":[],"_revision_number":1},{"id":"e4333ff78b955d69776ebc41cb748fd86cb93930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-14 09:05:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f15a617c13a0cc0fccd51ea1a7b9d0682c09086","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-14 09:07:17.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Update: https://review.opendev.org/c/openstack/cinderlib/+/768533 has been revised but has been stuck in the Zuul check all day.  Once it passes and has merged, either Rajat or I will update the hash here.\n\nhttps://review.opendev.org/c/openstack/cinderlib/+/768533 is merged and release patch is updated with the new hash","accounts_in_message":[],"_revision_number":2},{"id":"bb3242fea09f392938482ffb8283a30fc1d86019","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-14 09:18:51.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM thanks","accounts_in_message":[],"_revision_number":2},{"id":"17c96d51d1c38106e662eed3955637c45eec620f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 09:18:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464282c4ae624299957ad25498a957d9 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae5c94a7ee6f46b68e6b154d7a0754fe : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3c6531b9dcc4418c963ca642852b9753 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f0dc916375e4317d1f153262f7118ce98b1cd39c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-14 09:21:31.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, this patch has the hash to include the updated dependencies.","accounts_in_message":[],"_revision_number":2},{"id":"64fa2d7552e19f4537f977bed209379faf299df9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-14 13:07:45.000000000","message":"Patch Set 2: Code-Review+1\n\nVerified the hash, this is ready to go.  Thanks Gorka and Rajat!","accounts_in_message":[],"_revision_number":2},{"id":"bbf6a7556287301f72aaa2a64b8aacf1cf54ad3d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-14 13:12:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"784fd9705ce7751791d73cddd1b677c73b8dd7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:14:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e10fcf590219961cca3c321746a3886a1b537d8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:16:25.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"6b4ba3837cc536059ae13c7c6825a0a092ce8244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:40:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0268e58d0b5b4e97a4e0c210229e9767 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5195df39ccde4c419efa7476d26a05dd : SUCCESS in 8m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08d86c577dcc4078bb426bc744cf2830 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1d3ee4772299ca44885fb9568e29989921d3fe74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55eb54ccdb717328f419246130961157d4aa2a38","revisions":{"0f96ff0d4f10c84e28ae7a9e5f91a1128e53d572":{"kind":"REWORK","_number":1,"created":"2020-12-30 04:16:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/768766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/768766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/768766/1"}}},"commit":{"parents":[{"commit":"c13f9fc517a7301a8b1708dca73ea8be8029208e","subject":"Cinder team releases for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c13f9fc517a7301a8b1708dca73ea8be8029208e"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-30 04:14:01.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-30 04:14:01.000000000","tz":-300},"subject":"Release cinderlib 3.0.0","message":"Release cinderlib 3.0.0\n\nThis is the cycle-trailing Victoria release of cinderlib.\n\nChange-Id: I4c214340aab240427b8e3fa87fd91d8e688d92cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f96ff0d4f10c84e28ae7a9e5f91a1128e53d572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f96ff0d4f10c84e28ae7a9e5f91a1128e53d572"}]},"branch":"refs/heads/master"},"55eb54ccdb717328f419246130961157d4aa2a38":{"kind":"REWORK","_number":2,"created":"2021-01-14 09:05:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/768766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/768766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/768766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/768766/2"}}},"commit":{"parents":[{"commit":"c13f9fc517a7301a8b1708dca73ea8be8029208e","subject":"Cinder team releases for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c13f9fc517a7301a8b1708dca73ea8be8029208e"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-12-30 04:14:01.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-01-14 09:05:33.000000000","tz":-300},"subject":"Release cinderlib 3.0.0","message":"Release cinderlib 3.0.0\n\nThis is the cycle-trailing Victoria release of cinderlib.\n\nChange-Id: I4c214340aab240427b8e3fa87fd91d8e688d92cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55eb54ccdb717328f419246130961157d4aa2a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55eb54ccdb717328f419246130961157d4aa2a38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
