)]}'
{"id":"openstack%2Freleases~770896","triplet_id":"openstack%2Freleases~master~I75de9bd1b6d6a9bd2b9cca9ea349e8a489045cfb","project":"openstack/releases","branch":"master","topic":"w2-c-w-i","hashtags":[],"change_id":"I75de9bd1b6d6a9bd2b9cca9ea349e8a489045cfb","subject":"Wallaby-2 Release for python-mistralclient","status":"MERGED","created":"2021-01-15 06:48:13.000000000","updated":"2021-01-25 20:29:29.000000000","submitted":"2021-01-25 20:29:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770896-w2-c-w-i","meta_rev_id":"30170228cfec42dee4492c35338168c7496ae894","_number":770896,"virtual_id_number":770896,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 20:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2021-01-21 14:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2021-01-25 20:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 06:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-15 08:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 12:21:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"593baa90609b075970aff147ab82824a507bc219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-15 06:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54407d8c43ab6b4baaccd33614625df9450b882a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-15 06:57:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b997c810d4ba31dc77a00db225aa9e4b4e08d657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:12:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d224869496b547bdba940ca4b6325228 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2483afc37cb54b53a3d8de190a4759c5 : SUCCESS in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1e518798fc264489a676d2a7cf375d1a : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"aeb2a7202da89ec281a7dc7de6b7eb3a1f7de5f5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 14:45:13.000000000","message":"Patch Set 2: Code-Review+2\n\nOk deadline is today let\u0027s go ahead","accounts_in_message":[],"_revision_number":2},{"id":"7d13f8e1eaefc63b1b2a778dbf96d671e9b2348b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 20:17:05.000000000","message":"Patch Set 2: Workflow+1\n\nWell let\u0027s finish milestone 2","accounts_in_message":[],"_revision_number":2},{"id":"a01e20d5c8dd20f65306ef1f1730236a731d5888","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:17:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6266d42c2f44a4fcbddff99e9a4115d53064ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:29:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407e4c45bd69435481778047eb655256 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f59276f85184509b46514040b5c74ac : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eb3937eb62b40379ff4658c0baadfe6 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"30170228cfec42dee4492c35338168c7496ae894","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:29:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7328b8554af9337d2a4e6cf5522365b2531a5568","revisions":{"bfe9aa8611f9f57fea85c630c12777adf4c06891":{"kind":"REWORK","_number":1,"created":"2021-01-15 06:48:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/770896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/770896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/770896/1"}}},"commit":{"parents":[{"commit":"7e60245292cbbcdfb70e69729ff1c474f09425dc","subject":"Merge \"molteniron does not need releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e60245292cbbcdfb70e69729ff1c474f09425dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-15 06:48:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-15 06:48:04.000000000","tz":60},"subject":"Victoria-2 Release for python-mistralclient","message":"Victoria-2 Release for python-mistralclient\n\nThis is the Wallaby milestone-2 release for python-mistralclient.\nClients and libraries following the cycle-with-intermediary release\nmodel are encouraged to release each milestone to make sure updates\nare made available.\n\nIf you are happy with the current release hash, please +1 this patch\nand we will process a release. If you need to get some important\nchanges merged first, leave a -1 and update the patch with a new\ncommit hash to use once it is ready.\n\nIf a release is definitely not wanted at this time, please -1 and\nleave a comment explaining that and we can abandon the patch.\n\nAny patches with no response will be assumed to be OK and will be\nprocessed by January 21.\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I75de9bd1b6d6a9bd2b9cca9ea349e8a489045cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfe9aa8611f9f57fea85c630c12777adf4c06891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfe9aa8611f9f57fea85c630c12777adf4c06891"}]},"branch":"refs/heads/master"},"7328b8554af9337d2a4e6cf5522365b2531a5568":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-15 06:57:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/770896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/770896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/770896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/770896/2"}}},"commit":{"parents":[{"commit":"7e60245292cbbcdfb70e69729ff1c474f09425dc","subject":"Merge \"molteniron does not need releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e60245292cbbcdfb70e69729ff1c474f09425dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-15 06:48:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-15 06:57:44.000000000","tz":0},"subject":"Wallaby-2 Release for python-mistralclient","message":"Wallaby-2 Release for python-mistralclient\n\nThis is the Wallaby milestone-2 release for python-mistralclient.\nClients and libraries following the cycle-with-intermediary release\nmodel are encouraged to release each milestone to make sure updates\nare made available.\n\nIf you are happy with the current release hash, please +1 this patch\nand we will process a release. If you need to get some important\nchanges merged first, leave a -1 and update the patch with a new\ncommit hash to use once it is ready.\n\nIf a release is definitely not wanted at this time, please -1 and\nleave a comment explaining that and we can abandon the patch.\n\nAny patches with no response will be assumed to be OK and will be\nprocessed by January 21.\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I75de9bd1b6d6a9bd2b9cca9ea349e8a489045cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7328b8554af9337d2a4e6cf5522365b2531a5568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7328b8554af9337d2a4e6cf5522365b2531a5568"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
