)]}'
{"id":"openstack%2Freleases~772322","triplet_id":"openstack%2Freleases~master~I7c8399d13999980bf0ecf787b1c3f10bcf437745","project":"openstack/releases","branch":"master","topic":"update-owner","hashtags":[],"change_id":"I7c8399d13999980bf0ecf787b1c3f10bcf437745","subject":"Update the project owner","status":"MERGED","created":"2021-01-25 11:42:31.000000000","updated":"2021-02-17 13:04:37.000000000","submitted":"2021-02-17 13:04:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"772322-update-owner","meta_rev_id":"f59c3174e84a0d4ffd5078e52d19ceb6c3832971","_number":772322,"virtual_id_number":772322,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 13:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-17 12:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-02-16 12:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-17 12:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 11:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 11:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 14:37:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-02-05 14:37:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-02-08 13:07:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-02-09 10:18:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-02-11 14:19:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-11 15:04:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0b766ac97a8ae42155c2612417640e7e13bbbd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 11:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8775d0f57db71fa679b7e37c1403702be54d84bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 11:43:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b444a1a61e1aaf1c02fe2979019f0514b118ec77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 11:46:31.000000000","message":"Topic w2-c-w-i removed","accounts_in_message":[],"_revision_number":2},{"id":"208ba4cf6af88236436246a9a4cf07756ee409e6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 11:46:50.000000000","message":"Topic set to update-owner","accounts_in_message":[],"_revision_number":2},{"id":"aa49c993b8690d6e6a2c6dc906b33e271d363d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 11:52:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fface0d57348898166b5c34e224b67 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df592386c0164d9cbc6222458c5527c5 : SUCCESS in 4m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/01ff3795414e48778876099c27651c6a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"88abdc123aafe4d1cf95779d641856fa041e53a6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-05 15:10:43.000000000","message":"Patch Set 2:\n\nIf I understand correctly, Nova is now the owner of Placement\u0027s repositories. So my questions are:\n- don\u0027t we need to change \u0027team: placement-\u003enova\u0027 in deliverables/wallaby/placement.yaml too?\n- don\u0027t we need to change on every stable branches? (see the list below)\n\n$ grep \u0027team: placement\u0027 deliverables/*/*yaml\ndeliverables/_independent/os-resource-classes.yaml:team: placement\ndeliverables/_independent/os-traits.yaml:team: placement\ndeliverables/stein/os-resource-classes.yaml:team: placement\ndeliverables/stein/placement.yaml:team: placement\ndeliverables/train/placement.yaml:team: placement\ndeliverables/ussuri/osc-placement.yaml:team: placement\ndeliverables/ussuri/placement.yaml:team: placement\ndeliverables/victoria/osc-placement.yaml:team: placement\ndeliverables/victoria/placement.yaml:team: placement\ndeliverables/wallaby/osc-placement.yaml:team: placement\ndeliverables/wallaby/placement.yaml:team: placement","accounts_in_message":[],"_revision_number":2},{"id":"fbc36fc9a775442cc5e0ff8c2ecc1aea505c60e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 13:07:36.000000000","message":"Patch Set 2:\n\nI think we can probably keep the historical ones pointing to the separate placement team. Though now that I say that, I can\u0027t recall if that will cause us validation errors when try to do stable releases.","accounts_in_message":[],"_revision_number":2},{"id":"0e9dfa3b8d2eea3715c029c2874e5579e937733b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-08 13:38:16.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I think we can probably keep the historical ones pointing to the separate placement team. Though now that I say that, I can\u0027t recall if that will cause us validation errors when try to do stable releases.\n\nIf we want to avoid validation issues then that would mean we need to change the owner in branches: wallaby, victoria, ussuri and train.","accounts_in_message":[],"_revision_number":2},{"id":"72bba041dd41876557b319ca52c1d6f27bfb98ab","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-09 10:18:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nFast-tracking","accounts_in_message":[],"_revision_number":2},{"id":"ac6379d7e51fa67ccc31dbf6ad1ea48316845213","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-09 10:19:46.000000000","message":"Patch Set 2:\n\n(we should update the stable ones as well -- that entry is for validation, not history)","accounts_in_message":[],"_revision_number":2},{"id":"e1b7bdc32348738112d2807cf7bae538f2a1af2e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-09 10:20:19.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"effe273b52fa073cb41d9851dde49deade48f167","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-09 19:07:38.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e I think we can probably keep the historical ones pointing to the separate placement team. Though now that I say that, I can\u0027t recall if that will cause us validation errors when try to do stable releases.\n\u003e \n\u003e If we want to avoid validation issues then that would mean we need to change the owner in branches: wallaby, victoria, ussuri and train.\n\nYes good point, I\u0027ll push a new PS","accounts_in_message":[],"_revision_number":2},{"id":"e2230c247cd6558d1a82efdc9acbd1b1100b654a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-10 09:13:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4994b51d93b1d5e6c1417fd6d76926d0b8a1e65a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-10 09:14:49.000000000","message":"Patch Set 3:\n\nOnly W, V, U was erroneous, T, R, Q was already ok.","accounts_in_message":[],"_revision_number":3},{"id":"d313ed0e6406ff537d0181c61207c66a992077b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 09:22:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d769b736f745e5809dd821e610264f : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99ea3f32b1654862a9f9c814210c1948 : SUCCESS in 4m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/50cb46da52e943c0959a2bcf5ecfbf4b : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"e8410f82d83350b33a6ce132a0269566fe0e7717","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-10 10:16:51.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2bcebfd146ca55d3001c53cb24b555a5fc0755ec","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 14:19:22.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good as far as osc-placement is concerned. Placement itself could be added here or be done separately.","accounts_in_message":[],"_revision_number":3},{"id":"afa645fbdb0e55609ce2be11fe20cc11c6829dfd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-11 15:01:35.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+2\n\u003e \n\u003e Looks good as far as osc-placement is concerned. Placement itself could be added here or be done separately.\n\nOk I\u0027ll address that with a follow up patch","accounts_in_message":[],"_revision_number":3},{"id":"0223a42934af5b02786cc5ffc4e477b611500263","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-11 15:04:26.000000000","message":"Patch Set 3: Code-Review+1\n\nI\u0027m OK with the follow-up :)","accounts_in_message":[],"_revision_number":3},{"id":"03cc53417083778a6ba9283c8ec964e699564f7d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-11 15:07:51.000000000","message":"Patch Set 3:\n\nDone https://review.opendev.org/c/openstack/releases/+/775182","accounts_in_message":[],"_revision_number":3},{"id":"532fa9953cba5ef1803685b9c485ea65ab89cd79","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-16 12:02:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nOh, the follow-up already merged. Let\u0027s merge this one, too :)","accounts_in_message":[],"_revision_number":3},{"id":"3e667e868f27e369123827258be49028dd1da380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 12:02:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"82b1b55e947c81cafe18f0eaa10f55ba053f63c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 12:12:34.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"15d92c9fada16ad836d00e2c15cf47cf37ca7684","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-16 12:18:49.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks, now it replaces every occurrences of the former team name \u0027placement\u0027. LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"75eca93f8edde1385c6686005584224b80ab67a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:27:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d221637f89964682a848983252e754f6 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/890f3439293c472e93c87b04fbe5a6cf : SUCCESS in 4m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b6f85c299eb64a0c9bbdd1ee241b6a14 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":5},{"id":"ffff6b2b4c11ddf9066cace3882677b27f6ba2ef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-17 12:55:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d1eeecbf213cd3c0a244096c309d5cb1369356ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 12:55:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ef2d8f60d6bfb55454e275b5bf3fcc4861c44efe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 13:04:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af0c73755424822be43fd8c18b028da : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624971c0e524484d96974ee06f3a270b : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/422a0028362640158dd70395b038f7e4 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"f59c3174e84a0d4ffd5078e52d19ceb6c3832971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 13:04:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a56f80efc2a8085e04dfbb92a2043642725473ca","revisions":{"b831366951eeef7841f9fab8688bc14b6532c16f":{"kind":"REWORK","_number":1,"created":"2021-01-25 11:42:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/772322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/772322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/772322/1"}}},"commit":{"parents":[{"commit":"063b055db6ae88d1a0cb4adbabdbf56820f95e74","subject":"Merge \"Release OpenStack-Ansible Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/063b055db6ae88d1a0cb4adbabdbf56820f95e74"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"subject":"Update the project owner","message":"Update the project owner\n\nosc-placement have been moved to nova\n\nChange-Id: I7c8399d13999980bf0ecf787b1c3f10bcf437745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b831366951eeef7841f9fab8688bc14b6532c16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b831366951eeef7841f9fab8688bc14b6532c16f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"063b055db6ae88d1a0cb4adbabdbf56820f95e74","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"371a249c1161e0c0edbe7379234e5729ec73697c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-25 11:43:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/772322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/772322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/772322/2"}}},"commit":{"parents":[{"commit":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","subject":"Wallaby-2 Release for osc-placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/465504eab3ce0dc0e3cedcdf54acb0b256d09118"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:43:01.000000000","tz":60},"subject":"Update the project owner","message":"Update the project owner\n\nosc-placement have been moved to nova\n\nChange-Id: I7c8399d13999980bf0ecf787b1c3f10bcf437745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/371a249c1161e0c0edbe7379234e5729ec73697c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/371a249c1161e0c0edbe7379234e5729ec73697c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","is_merged_in_target_branch":true,"change_id":"I12e6a9232b0dc7c7ae325cb7394a8923e4733902","change_number":770887,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d5fbd5f282d75e91b83c4fd427e19bdca48369fd":{"kind":"REWORK","_number":3,"created":"2021-02-10 09:13:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/772322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/772322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/772322/3"}}},"commit":{"parents":[{"commit":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","subject":"Wallaby-2 Release for osc-placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/465504eab3ce0dc0e3cedcdf54acb0b256d09118"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-10 09:12:59.000000000","tz":60},"subject":"Update the project owner","message":"Update the project owner\n\nosc-placement have been moved to nova\n\nChange-Id: I7c8399d13999980bf0ecf787b1c3f10bcf437745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5fbd5f282d75e91b83c4fd427e19bdca48369fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5fbd5f282d75e91b83c4fd427e19bdca48369fd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","is_merged_in_target_branch":true,"change_id":"I12e6a9232b0dc7c7ae325cb7394a8923e4733902","change_number":770887,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8065c1c2896573dc768acd431d253d54c4a5ce9d":{"kind":"REWORK","_number":4,"created":"2021-02-16 12:02:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/772322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/772322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/772322/4"}}},"commit":{"parents":[{"commit":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","subject":"Wallaby-2 Release for osc-placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/465504eab3ce0dc0e3cedcdf54acb0b256d09118"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-16 12:01:39.000000000","tz":60},"subject":"Update the project owner","message":"Update the project owner\n\nplacement projects have been moved to nova\n\nChange-Id: I7c8399d13999980bf0ecf787b1c3f10bcf437745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8065c1c2896573dc768acd431d253d54c4a5ce9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8065c1c2896573dc768acd431d253d54c4a5ce9d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"465504eab3ce0dc0e3cedcdf54acb0b256d09118","is_merged_in_target_branch":true,"change_id":"I12e6a9232b0dc7c7ae325cb7394a8923e4733902","change_number":770887,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a56f80efc2a8085e04dfbb92a2043642725473ca":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-16 12:12:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/772322/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/772322/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/772322/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/772322/5"}}},"commit":{"parents":[{"commit":"71c178a219379616f167d4a09c0e03be29524f21","subject":"Merge \"Horizon stable releases (Ussuri, Victoria)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71c178a219379616f167d4a09c0e03be29524f21"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 11:41:57.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-16 12:12:19.000000000","tz":60},"subject":"Update the project owner","message":"Update the project owner\n\nplacement projects have been moved to nova\n\nChange-Id: I7c8399d13999980bf0ecf787b1c3f10bcf437745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56f80efc2a8085e04dfbb92a2043642725473ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56f80efc2a8085e04dfbb92a2043642725473ca"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"71c178a219379616f167d4a09c0e03be29524f21","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
