)]}'
{"id":"openstack%2Freleases~772570","triplet_id":"openstack%2Freleases~master~I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e","project":"openstack/releases","branch":"master","topic":"I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e","hashtags":[],"change_id":"I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e","subject":"Move os-apply|collect|refresh-config projects to independent","status":"MERGED","created":"2021-01-26 16:39:03.000000000","updated":"2021-02-08 13:16:25.000000000","submitted":"2021-02-08 13:16:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"772570-I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e","meta_rev_id":"38e36a787ced981caee572c87baa29f84ca40bcc","_number":772570,"virtual_id_number":772570,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 13:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 13:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-01-29 13:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-02-04 16:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 13:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 16:52:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-01-26 16:52:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-01-26 16:52:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-26 16:52:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-01-26 16:52:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"state":"REVIEWER"},{"updated":"2021-01-26 16:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 17:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-01-27 09:11:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 10:18:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-02-08 13:04:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"65a4fb565e848b51a0e8dd68c60916c45c27ec72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-26 16:39:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a12e96e3549b6ecb9b5c27874231177160915a42","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:54:22.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"659ce386c9fa36ac9e86783c416c179cb2fb0882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:25:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0004859a87c74362b09842224a1d2cb5 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71b188dda6934a61a6b5f108ecafc8a0 : SUCCESS in 4m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4bccd58974f4402e8997630d0125715c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"493bc7e1eb166e29d005b237a88b847f67562a6b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-26 17:38:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"990f4d7d8bfc7799553e7de41cfa931116bcb339","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-27 09:11:09.000000000","message":"Patch Set 1: Code-Review-1\n\nHello,\n\nThanks.\n\nWe need to remove the wallaby files:\n\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-refresh-config.yaml\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-collect-config.yaml\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-apply-config.yaml","accounts_in_message":[],"_revision_number":1},{"id":"93f9518bc295b00d8aef0b34712a913953f7a5ec","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-27 15:02:38.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Hello,\n\u003e \n\u003e Thanks.\n\u003e \n\u003e We need to remove the wallaby files:\n\u003e \n\u003e - https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-refresh-config.yaml\n\u003e - https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-collect-config.yaml\n\u003e - https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/os-apply-config.yaml\n\nthanks doing now\n\ndepending on whether this or https://review.opendev.org/c/openstack/releases/+/771830 merges first, I will update the other.","accounts_in_message":[],"_revision_number":1},{"id":"589928b124de8c1a9a592b1e7954de781833cbd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-27 15:02:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e55e8e2d4c7e6b257d81edd5b7b214ca4f31f38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:17:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b191c317645f4c509491039e6a6e09e0 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8123bffa3c474106b47d42af94d9bb45 : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/90d5f29678404dda8b188dc8352825dd : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a00504d2a7ed295af2656f9f6073a0f0f55baaba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-27 15:19:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f16f4fee01badb305e2029b81210e65fefee1c1","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-01-28 09:56:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d81c53a7738f452be8fe728b18e92d3ba84e431a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-01-28 10:18:52.000000000","message":"Patch Set 2:\n\nI don\u0027t think we should mention older releases (from victoria and before) in the _independent file. That\u0027s duplicating history and a bit confusing... We can discuss it at the meeting later today.","accounts_in_message":[],"_revision_number":2},{"id":"7db9ae8b3efb78c785b3ec225d7f0cd85cee55c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:43:32.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I don\u0027t think we should mention older releases (from victoria and before) in the _independent file. That\u0027s duplicating history and a bit confusing... We can discuss it at the meeting later today.\n\n+1","accounts_in_message":[],"_revision_number":2},{"id":"568bad4a2313b163858d85485dca5abe438a1d84","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 14:07:49.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I don\u0027t think we should mention older releases (from victoria and before) in the _independent file. That\u0027s duplicating history and a bit confusing... We can discuss it at the meeting later today.\n\nOK thanks, i wasn\u0027t sure if that was the right thing to do or not. I will update it","accounts_in_message":[],"_revision_number":2},{"id":"93a47d7ea33aa08021f569f777282aae75fcb8f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 14:12:58.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e I don\u0027t think we should mention older releases (from victoria and before) in the _independent file. That\u0027s duplicating history and a bit confusing... We can discuss it at the meeting later today.\n\u003e \n\u003e OK thanks, i wasn\u0027t sure if that was the right thing to do or not. I will update it\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"12ca126829382380327f910b0f4ff6429a1b31f9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 14:13:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b64fd9edd21dc0b12b827b3cb0762b5869422a4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 14:33:27.000000000","message":"Patch Set 2:\n\nis there some ongoing work with the validators i wonder.\n\nwith release-model: cycle-with-intermediary\n\ntox validate fails with \ndeliverables/_independent/os-apply-config.yaml: validate_model: deliverables in the _independent directory should use either the independent or abandoned release models\n\nwhen i use release-model: independent \n\ntox validate fails with \nFailed validating \u0027enum\u0027 in schema[\u0027properties\u0027][\u0027release-model\u0027]:\n    {\u0027enum\u0027: [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027],\n     \u0027type\u0027: \u0027string\u0027}\n\nOn instance[\u0027release-model\u0027]:\n    \u0027independent\u0027\ndeliverables/_independent/os-collect-config.yaml: \u0027independent\u0027 is not one of [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027]","accounts_in_message":[],"_revision_number":2},{"id":"8212f29984b8ee259dce9e187e88bc002ae4f546","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 14:34:29.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e is there some ongoing work with the validators i wonder.\n\u003e \n\u003e with release-model: cycle-with-intermediary\n\u003e \n\u003e tox validate fails with \n\u003e deliverables/_independent/os-apply-config.yaml: validate_model: deliverables in the _independent directory should use either the independent or abandoned release models\n\u003e \n\u003e when i use release-model: independent \n\u003e \n\u003e tox validate fails with \n\u003e Failed validating \u0027enum\u0027 in schema[\u0027properties\u0027][\u0027release-model\u0027]:\n\u003e     {\u0027enum\u0027: [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027],\n\u003e      \u0027type\u0027: \u0027string\u0027}\n\u003e \n\u003e On instance[\u0027release-model\u0027]:\n\u003e     \u0027independent\u0027\n\u003e deliverables/_independent/os-collect-config.yaml: \u0027independent\u0027 is not one of [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027]\n\n\n\nI am going to use independent in my update even though the validation will fail and I will update as needed thanks","accounts_in_message":[],"_revision_number":2},{"id":"afc369d7226ea080f78a568076919925fad87ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 14:36:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ef27392cc805e41533773b787a73991b6b4a4cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 14:49:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa02612f1a341a9a0b4e554b28c81cd : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/022430d48e5a4d07b8f4bae20ea93d8d : FAILURE in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/30c5e3df95c5484a8954d8fafb3af30b : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"51f91b715336ce37d671905d07026079d2950a3a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 15:13:47.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e is there some ongoing work with the validators i wonder.\n\u003e \u003e \n\u003e \u003e with release-model: cycle-with-intermediary\n\u003e \u003e \n\u003e \u003e tox validate fails with \n\u003e \u003e deliverables/_independent/os-apply-config.yaml: validate_model: deliverables in the _independent directory should use either the independent or abandoned release models\n\u003e \u003e \n\u003e \u003e when i use release-model: independent \n\u003e \u003e \n\u003e \u003e tox validate fails with \n\u003e \u003e Failed validating \u0027enum\u0027 in schema[\u0027properties\u0027][\u0027release-model\u0027]:\n\u003e \u003e     {\u0027enum\u0027: [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027],\n\u003e \u003e      \u0027type\u0027: \u0027string\u0027}\n\u003e \u003e \n\u003e \u003e On instance[\u0027release-model\u0027]:\n\u003e \u003e     \u0027independent\u0027\n\u003e \u003e deliverables/_independent/os-collect-config.yaml: \u0027independent\u0027 is not one of [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027]\n\u003e \n\u003e \n\u003e \n\u003e I am going to use independent in my update even though the validation will fail and I will update as needed thanks\n\nThe `release-model` (key/value) is for coordinated series, independent deliverables don\u0027t need that.\nYou can remove that line.","accounts_in_message":[],"_revision_number":3},{"id":"b926e7f18b07a8c125ebf8f68bd296d3af764c8a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 17:40:32.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e is there some ongoing work with the validators i wonder.\n\u003e \u003e \u003e \n\u003e \u003e \u003e with release-model: cycle-with-intermediary\n\u003e \u003e \u003e \n\u003e \u003e \u003e tox validate fails with \n\u003e \u003e \u003e deliverables/_independent/os-apply-config.yaml: validate_model: deliverables in the _independent directory should use either the independent or abandoned release models\n\u003e \u003e \u003e \n\u003e \u003e \u003e when i use release-model: independent \n\u003e \u003e \u003e \n\u003e \u003e \u003e tox validate fails with \n\u003e \u003e \u003e Failed validating \u0027enum\u0027 in schema[\u0027properties\u0027][\u0027release-model\u0027]:\n\u003e \u003e \u003e     {\u0027enum\u0027: [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027],\n\u003e \u003e \u003e      \u0027type\u0027: \u0027string\u0027}\n\u003e \u003e \u003e \n\u003e \u003e \u003e On instance[\u0027release-model\u0027]:\n\u003e \u003e \u003e     \u0027independent\u0027\n\u003e \u003e \u003e deliverables/_independent/os-collect-config.yaml: \u0027independent\u0027 is not one of [\u0027cycle-with-intermediary\u0027, \u0027cycle-with-milestones\u0027, \u0027cycle-trailing\u0027, \u0027untagged\u0027, \u0027cycle-with-rc\u0027, \u0027cycle-automatic\u0027, \u0027abandoned\u0027]\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e \n\u003e \u003e I am going to use independent in my update even though the validation will fail and I will update as needed thanks\n\u003e \n\u003e The `release-model` (key/value) is for coordinated series, independent deliverables don\u0027t need that.\n\u003e You can remove that line.\n\nthank you very much Herve updating","accounts_in_message":[],"_revision_number":3},{"id":"0747b4ad05a6b62abc7ce82eea8436fee512ba31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-28 17:40:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"544b70099542651b5071846145cd6b8be28a4b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:48:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e407a493dcae402cb72c15def9e66041 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b12830fa68914ae78fa388ef54ce45de : SUCCESS in 4m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2b2de45902c64fb1b8283cc6e11e0aea : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":4},{"id":"02776b0d174f95fee8fbec27c6506dade988aa1f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-29 13:56:51.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f6708abfa2fb0b87a4541e8588575cb25d333b0a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 12:44:06.000000000","message":"Patch Set 4: Code-Review+2\n\nHmm I think we should tweak some data because the PTL/liaison validation seems broken, I\u0027ll take a look","accounts_in_message":[],"_revision_number":4},{"id":"f89afc0644ef422281ac296a822be2533fede422","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 15:35:37.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"917bf8cb57741808aab61e29df18d86807a886c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 16:02:55.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e Patch Set 4: Code-Review+2\n\u003e \n\u003e Hmm I think we should tweak some data because the PTL/liaison validation seems broken, I\u0027ll take a look\n\nOk I got it, the PLT check failed because the removed the os-collect-config file and I think that the validation is based on list files that have been updated (`git show -n 1 --pretty\u003d\"\" --name-only\n`), so the check try to validate a removed file...\n\nhttps://zuul.opendev.org/t/openstack/build/240d6ce1f3074163b4e56cf9d044e60d","accounts_in_message":[],"_revision_number":4},{"id":"3b322e07c07c8607c3b0f13f82033516a1d2fec3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 13:04:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nMakes sense, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"658fa307eb479c9f3b441535f7595de3dd8fd862","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:04:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"df6f0821319891c2d6dd4c034a89fbcbdc5cf620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:16:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41818869fdc74023b65b97c6ee0678c2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ffd8bc3d4b46cd8bb778dfda3a5634 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16873443288f4b66bf532cb839983f28 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"38e36a787ced981caee572c87baa29f84ca40bcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:16:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78d0577401838b7066341572346f7e7e99640d40","revisions":{"2d96ef58619755e00a49f377eb825cfe0b84455c":{"kind":"REWORK","_number":1,"created":"2021-01-26 16:39:03.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/70/772570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/772570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/772570/1"}}},"commit":{"parents":[{"commit":"eb381492da3f7c826c35b9f147fd9a1ed55ae797","subject":"Merge \"Wallaby-2 Release for python-muranoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb381492da3f7c826c35b9f147fd9a1ed55ae797"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-26 16:35:46.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-26 16:35:46.000000000","tz":120},"subject":"Move os-apply|collect|refresh-config projects to independent","message":"Move os-apply|collect|refresh-config projects to independent\n\nAs discussed at [1] this proposes that the os-apply-config\nos-refresh-config and os-collect-config repos are moved\nto the independent model.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019778.html\n\nChange-Id: I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d96ef58619755e00a49f377eb825cfe0b84455c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d96ef58619755e00a49f377eb825cfe0b84455c"}]},"branch":"refs/heads/master"},"f92837169a3fc8be0694a8415e1091ce6f537925":{"kind":"REWORK","_number":2,"created":"2021-01-27 15:02:58.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/70/772570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/772570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/772570/2"}}},"commit":{"parents":[{"commit":"eb381492da3f7c826c35b9f147fd9a1ed55ae797","subject":"Merge \"Wallaby-2 Release for python-muranoclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb381492da3f7c826c35b9f147fd9a1ed55ae797"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-26 16:35:46.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-27 15:00:51.000000000","tz":120},"subject":"Move os-apply|collect|refresh-config projects to independent","message":"Move os-apply|collect|refresh-config projects to independent\n\nAs discussed at [1] this proposes that the os-apply-config\nos-refresh-config and os-collect-config repos are moved\nto the independent model. Removes the wallaby release files\nfor these repos.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019778.html\n\nChange-Id: I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f92837169a3fc8be0694a8415e1091ce6f537925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f92837169a3fc8be0694a8415e1091ce6f537925"}]},"branch":"refs/heads/master"},"b14f66a2b94fa882083064dc61d7ed8e854f4868":{"kind":"REWORK","_number":3,"created":"2021-01-28 14:36:27.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/70/772570/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/772570/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/772570/3"}}},"commit":{"parents":[{"commit":"ff13096d8054c34b9ccc74c22e8d4e41dab5f65b","subject":"Merge \"[Tempest Plugins] Tag stein-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff13096d8054c34b9ccc74c22e8d4e41dab5f65b"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-26 16:35:46.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-28 14:29:23.000000000","tz":120},"subject":"Move os-apply|collect|refresh-config projects to independent","message":"Move os-apply|collect|refresh-config projects to independent\n\nAs discussed at [1] this proposes that the os-apply-config\nos-refresh-config and os-collect-config repos are moved\nto the independent model. Removes the wallaby release files\nfor these repos.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019778.html\n\nChange-Id: I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b14f66a2b94fa882083064dc61d7ed8e854f4868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b14f66a2b94fa882083064dc61d7ed8e854f4868"}]},"branch":"refs/heads/master"},"78d0577401838b7066341572346f7e7e99640d40":{"kind":"REWORK","_number":4,"created":"2021-01-28 17:40:42.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/70/772570/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/772570/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/772570/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/772570/4"}}},"commit":{"parents":[{"commit":"ff13096d8054c34b9ccc74c22e8d4e41dab5f65b","subject":"Merge \"[Tempest Plugins] Tag stein-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff13096d8054c34b9ccc74c22e8d4e41dab5f65b"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-26 16:35:46.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-28 17:39:15.000000000","tz":120},"subject":"Move os-apply|collect|refresh-config projects to independent","message":"Move os-apply|collect|refresh-config projects to independent\n\nAs discussed at [1] this proposes that the os-apply-config\nos-refresh-config and os-collect-config repos are moved\nto the independent model. Removes the wallaby release files\nfor these repos.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019778.html\n\nChange-Id: I6c4ab66ac71f9e6050b0944f15b9fef7fccfe89e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78d0577401838b7066341572346f7e7e99640d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78d0577401838b7066341572346f7e7e99640d40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
