)]}'
{"id":"openstack%2Freleases~772845","triplet_id":"openstack%2Freleases~master~Ifcfd1b244c3a7e9d26b762585bab61bd43e28a9b","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ifcfd1b244c3a7e9d26b762585bab61bd43e28a9b","subject":"Second try to release.","status":"ABANDONED","created":"2021-01-28 09:51:50.000000000","updated":"2021-03-09 11:39:22.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"d249a63053634e8a7a9bcccede7c6487d0f20707","_number":772845,"virtual_id_number":772845,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-28 10:02:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2021-01-28 10:32:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-01-28 09:54:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 09:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 10:32:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-01-28 10:32:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"4bf44ab9aba8ea1487c896c6ccef94f1e7cffc64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-28 09:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda8359ba147d0296332b3a972f8850681ba03bc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 09:54:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"32d976312e8ead0c6911ce79e9afb877d7638897","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-28 09:55:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c41880087a9f3e776025993e34c821e3d017877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7737a7c079394e33ab09b5e1511bc951 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a2125ba5f4a417ab898c5b774887c61 : FAILURE in 4m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b220e45c30094749881819ae643eecfc : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d0133ef3bbee665ae00f02de23cb50399c400c4e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 10:32:07.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHello Martin\n\nThanks for these changes, see my inline comment.","accounts_in_message":[],"_revision_number":1},{"id":"87492b9ca28ba8fc897d9455c43a35954f18449b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 10:32:16.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"c32c8dbedd61be8213ff1ff5cb17716eb37746b2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 10:38:12.000000000","message":"Topic tempest-plugin-stein-last removed","accounts_in_message":[],"_revision_number":1},{"id":"e577a760eb5b35fec2cfb276beca04cd767244c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 10:41:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ac2a3ea2519e0770dc891eb38a1cf87257d2c9f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-05 13:12:15.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"7f5395c844082f6a676272129805b72513b6ed74","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-02 09:16:58.000000000","message":"Patch Set 1:\n\nHello, any updates?","accounts_in_message":[],"_revision_number":1},{"id":"210e972c149d2e253c5563f06d2764953fbe41ab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 08:03:51.000000000","message":"Patch Set 1:\n\nHello,\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"b24d6360105eff18204f6478cfedaa525a431cdf","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-09 11:38:38.000000000","message":"Patch Set 1:\n\nHi Hervé,\nI went through this topic again:\n\n1. The Tag 2.2.0 was created successfully [1]\n2. After the merge the Post-job publish-monasca-tempest-plugin-docker-image\nfailed because of a change in a new docker version that requires to add the\ntag to push images into Docker Hub. That bug is already fixed. Our periodic\npush job is working well now [2].\n3. Today I built the image manually from Tag 2.2.0 and pushed into Monasca\u0027s\nDocker Hub account [3]\n\nThen I abandon this change here.\n\nSorry for late answer,\nMartin (chaconpiza) \n\n[1] https://opendev.org/openstack/monasca-tempest-plugin/src/tag/2.2.0\n[2] https://zuul.opendev.org/t/openstack/builds?job_name\u003dpublish-monasca-tempest-plugin-docker-image\n[3] https://hub.docker.com/layers/monasca/tempest-tests/2.2.0/images/sha256-5be4bd6f0ee738d033cfedf7fae4b122907300a481d37187262c50ef56e48576?context\u003dexplore","accounts_in_message":[],"_revision_number":1},{"id":"d249a63053634e8a7a9bcccede7c6487d0f20707","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-09 11:39:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab9742f4f607f737b6489fc3e96a7d4426230440","revisions":{"ab9742f4f607f737b6489fc3e96a7d4426230440":{"kind":"REWORK","_number":1,"created":"2021-01-28 09:51:50.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/45/772845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/772845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/772845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/772845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/772845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/772845/1"}}},"commit":{"parents":[{"commit":"d0550db588c787cef00f978559e46654d10754bf","subject":"Merge \"Release XStatic-JQuery-Migrate-3.3.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0550db588c787cef00f978559e46654d10754bf"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-01-28 09:48:21.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-01-28 09:48:21.000000000","tz":60},"subject":"Second try to release.","message":"Second try to release.\n\nFirst can be tracked on [3]\n\n[Tempest Plugins] Tag stein-last\n\nStein branch is in Extended Maintenance now[1]. All Tempest\nplugins are branchless which means master version of Tempest\nand its plugins is used to test the supported stable branches.\n\nOnce stable branch is moved to EM state then, Tempest and its\nplugins compatible tag needs to be released so that we can\nkeep testing the EM stable branches with this tag once master\nTempest and its plugins are not compatible[2].\n\nThis tag will help for testing the stable/stein with single\ncompatible tag of Tempest and its plugins.\n\nTagging \u0027stein-last\u0027 as well as a new version also with same hash.\nDocumenting this new tag name here\n- https://review.opendev.org/c/openstack/project-team-guide/+/769821\n\nMore detail on compatible version policy for Tempest and its\nplugins is here\n- https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\n[1] https://releases.openstack.org/stein/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://review.opendev.org/c/openstack/releases/+/768551\n\nChange-Id: Ifcfd1b244c3a7e9d26b762585bab61bd43e28a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab9742f4f607f737b6489fc3e96a7d4426230440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab9742f4f607f737b6489fc3e96a7d4426230440"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
