)]}'
{"id":"openstack%2Freleases~773070","triplet_id":"openstack%2Freleases~master~Id029fed75a72858eec9d983bb015007c144c96b3","project":"openstack/releases","branch":"master","topic":"train-stable","hashtags":[],"change_id":"Id029fed75a72858eec9d983bb015007c144c96b3","subject":"Release networking-ovn for stable/train","status":"MERGED","created":"2021-01-29 14:28:37.000000000","updated":"2021-03-10 14:42:42.000000000","submitted":"2021-03-10 14:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773070-train-stable","meta_rev_id":"d368f56a42fb7c47ea253f39b9d94fdc976bbfb4","_number":773070,"virtual_id_number":773070,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 14:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-10 12:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-03-10 12:16:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-03-10 14:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-03-10 14:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-10 12:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 14:30:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 14:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 15:23:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-01-29 15:23:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-29 18:40:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-08 10:33:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-03-10 14:20:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f03c92a47ac062b1595f342d1f89d3b7d32dabdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-29 14:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914cdc849c4fc72cd2cd7e23c8c2eacaad3fa901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 14:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d4ddae27c1471eac760b2acac5cd99 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03f31f18c1174446af731e9c06abf5c4 : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d0c8065f1c25455fb1f51bd7efcf26f2 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"37949500358475168fef24ef184680163b822149","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-29 18:40:00.000000000","message":"Patch Set 1: Code-Review-1\n\nThere are a few more reviews in-flight, let\u0027s wait another day or two.  I can update this with the new SHA then.","accounts_in_message":[],"_revision_number":1},{"id":"3e5aecdccf60a9d6156af73cca7175c28e630306","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-29 20:21:32.000000000","message":"Patch Set 1:\n\nThanks Brian for the feedback, no need to hurry, take your time with the release. :)","accounts_in_message":[],"_revision_number":1},{"id":"2f92468c19a1c00eda73601b8d4c0fbe39fe3ad7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 15:06:59.000000000","message":"Patch Set 1:\n\n@Brian are there any news regarding networking-ovn train release? Train transition to Extended Maintenance is closer and closer and maybe it is best to roll out sooner \u0026 smaller releases. Anyway, let me know if this patch is still valid or you want to take it over or we should abandon this. Thanks :)","accounts_in_message":[],"_revision_number":1},{"id":"72fc0208b49d15cb54f0a891c459d9fcc4b00242","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-26 16:35:48.000000000","message":"Patch Set 1:\n\nHi Elod, I was waiting for our train queue to drain to zero but one change is stuck in review.  I\u0027ll update with a new sha Monday just so we can get something out.  Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"fa92f48b6469c5455ed44e0ea406c3219914c76a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 10:33:12.000000000","message":"Patch Set 1:\n\nHello,\n\nAny updates?\n\nI seen that 3 patches (I ignored the \"DNM\" patch) remains unmerged on train https://review.opendev.org/q/project:openstack/networking-ovn+branch:stable/train+status:open\n\nDo we want to wait for all of them?","accounts_in_message":[],"_revision_number":1},{"id":"529eb78ad0263d0c5675a180db278a92dca0df87","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 10:50:52.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Hello,\n\u003e \n\u003e Any updates?\n\u003e \n\u003e I seen that 3 patches (I ignored the \"DNM\" patch) remains unmerged on train https://review.opendev.org/q/project:openstack/networking-ovn+branch:stable/train+status:open\n\u003e \n\u003e Do we want to wait for all of them?\n\nSorry, I missed that patch. I just approved https://review.opendev.org/c/openstack/networking-ovn/+/777224 and rechecked https://review.opendev.org/c/openstack/networking-ovn/+/772094\n\nFrom the Brian\u0027s comment in https://review.opendev.org/c/openstack/networking-ovn/+/777030 I don\u0027t think we need to wait for it but I asked him there to be sure about it.","accounts_in_message":[],"_revision_number":1},{"id":"948081cb9c9f648a7fb8baad04360875f45a2b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-08 18:06:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"378cb5e1b5e387e0ec006f26ad377b3b4fc815f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-08 18:07:59.000000000","message":"Patch Set 2:\n\nI updated the hash to the latest.\n\nThe only change that might merge this week is https://review.opendev.org/c/openstack/networking-ovn/+/777224 - we can always just do another release later since there are other patches still pending.","accounts_in_message":[],"_revision_number":2},{"id":"a233fbc76fd22b0225430635fdf26b5dbeb293be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 18:15:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/117db02b5962405aa82e92ed98c43698 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a3a597ebd674468876fd279a7439106 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f397a96481a644378c4ed4efe202b44b : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f91662c85d21ad18261efbb067413774a3c9ba4a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 22:12:02.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e Patch Set 2:\n\u003e \n\u003e I updated the hash to the latest.\n\u003e \n\u003e The only change that might merge this week is https://review.opendev.org/c/openstack/networking-ovn/+/777224 - we can always just do another release later since there are other patches still pending.\n\nBased on that comment from Brian, I\u0027m ok with that now.","accounts_in_message":[],"_revision_number":2},{"id":"054228e89ed786ddbc4211cc2dc4c83bc1f29b6c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:12:59.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"4c9bc4a1c4b1a5f1c06d332cb4fba17b07053092","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 08:38:12.000000000","message":"Patch Set 2:\n\nRebased to train\u0027s HEAD.\n\nRuntime requirements have been updated so it could be worth to propose a minor version to highlight this point.","accounts_in_message":[],"_revision_number":2},{"id":"f318a2a61287b44f6a6da9042b62cdfdff512664","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-09 10:41:52.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Rebased to train\u0027s HEAD.\n\u003e \n\u003e Runtime requirements have been updated so it could be worth to propose a minor version to highlight this point.\n\nImpressive amount of newly merged patches :) I also thought about minor version bump as there are some changes that could also involve that (e.g. patch: 503921ca 2021-01-18 11:52:41 +0000 ovn: Support live migration to DPDK nodes ?), but I might be wrong. Any feeling about this?\n\n(Otherwise the release should be OK.)","accounts_in_message":[],"_revision_number":2},{"id":"30e98b71a3d1744bf0e5252df181c9be530d0b9e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 11:08:29.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e Rebased to train\u0027s HEAD.\n\u003e \u003e \n\u003e \u003e Runtime requirements have been updated so it could be worth to propose a minor version to highlight this point.\n\u003e \n\u003e Impressive amount of newly merged patches :) I also thought about minor version bump as there are some changes that could also involve that (e.g. patch: 503921ca 2021-01-18 11:52:41 +0000 ovn: Support live migration to DPDK nodes ?), but I might be wrong. Any feeling about this?\n\u003e \n\u003e (Otherwise the release should be OK.)\n\nI was also thinking about it yesterday. So if You both think it would be better, let\u0027s change it :)","accounts_in_message":[],"_revision_number":2},{"id":"6d3e34a470993174285afbc6dee6abbbbc7056cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 11:09:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d3fa6ea1a14337bdb8d31729d1f95a948c3ddccc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 11:09:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5638d5c35bdaa38e27fce5f944c5fc5305949a3e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 11:10:19.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dc3d336861b0f350be80925abf9e021c6babffc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 11:18:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6933a70867a4ad69ab20ba015d722ae : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d09a9367ab64905a6e51f7a8ce619e6 : SUCCESS in 7m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f2bded24e3247aaaac53bf0f544511e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"37ebae22819edee52c0913adfa82ab2e89b7c28d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-09 12:48:48.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM. Thanks Slawek! :)","accounts_in_message":[],"_revision_number":3},{"id":"97bb70300cd708dead4f75d2951c08e4412c1ec4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-09 17:04:03.000000000","message":"Patch Set 3: Code-Review-1\n\nCan we wait for https://review.opendev.org/c/openstack/networking-ovn/+/777224 ?  It just entered the gate since the get-pip issue was fixed.  I\u0027ll update in a little bit once it merges.","accounts_in_message":[],"_revision_number":3},{"id":"808a395b434125be5444cac547c39c5cc2a3c323","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-09 18:01:02.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e Can we wait for https://review.opendev.org/c/openstack/networking-ovn/+/777224 ?  It just entered the gate since the get-pip issue was fixed.  I\u0027ll update in a little bit once it merges.\n\nNo problem Brian, if that is on the way, then of course it worth to wait for :)","accounts_in_message":[],"_revision_number":3},{"id":"ff5116803222f153186845be0b9a5c8dd22bcc89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-09 20:00:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cdd83512412498919ed9761d99081f6482aba08d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-09 20:01:06.000000000","message":"Patch Set 4:\n\nThis should be good now, thanks.","accounts_in_message":[],"_revision_number":4},{"id":"e71fe1e6a9fad183dfc290fa002cac85487d6d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:08:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74d32b2cd114140b884ee7dc31583c3 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a83e39640954ac388e989baa26154d8 : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5d148ad68fc444bf84ff2cb99a8b133f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":4},{"id":"d2d629d3c6263e552e61a664f276564b32ba9dcf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-10 12:04:54.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks for the update Brian, the release patch looks OK to me.","accounts_in_message":[],"_revision_number":4},{"id":"2f81e728e72cc813cbfdbe9482fc382a7804bd69","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-10 12:16:24.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Brian for taking care of it. LGTM.","accounts_in_message":[],"_revision_number":4},{"id":"2aad906d36f7f2a36d94ff488100fd9db3b71bbd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:17:57.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"483af7e28ff4f1b87254bd5350d050c3eee6959a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-10 14:20:11.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM waiting for PTL approval","accounts_in_message":[],"_revision_number":4},{"id":"40b458a72b0319f93eb1d3f7c35887367eb427fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-10 14:27:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2c8b0823148f7ceda94a97f1757aaddbe761c7b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:28:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"005e90400ecc3c7fdbcfb0d9e48ee6c58ddc04d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:42:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d95ad4d0cc4ac3b0811d9a3c36527f : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627aa83853c64cd285efdd7ca73b01e4 : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69b7f3e8d79a4f5991175f8e9e30cd66 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"d368f56a42fb7c47ea253f39b9d94fdc976bbfb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:42:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e456493b254f67885225854309916643c0e467a","revisions":{"5baf925d03727703f8052620192b92f14d238090":{"kind":"REWORK","_number":1,"created":"2021-01-29 14:28:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/70/773070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/773070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/773070/1"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:28:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:28:31.000000000","tz":60},"subject":"Release networking-ovn for stable/train","message":"Release networking-ovn for stable/train\n\nThis release picks up new commits to networking-ovn since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id029fed75a72858eec9d983bb015007c144c96b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5baf925d03727703f8052620192b92f14d238090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5baf925d03727703f8052620192b92f14d238090"}]},"branch":"refs/heads/master"},"b3c7f1ea5ff3d1d20de093d6cf8144c450bba659":{"kind":"REWORK","_number":2,"created":"2021-03-08 18:06:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/773070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/773070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/773070/2"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:28:31.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-08 18:05:34.000000000","tz":-300},"subject":"Release networking-ovn for stable/train","message":"Release networking-ovn for stable/train\n\nThis release picks up new commits to networking-ovn since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id029fed75a72858eec9d983bb015007c144c96b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3c7f1ea5ff3d1d20de093d6cf8144c450bba659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3c7f1ea5ff3d1d20de093d6cf8144c450bba659"}]},"branch":"refs/heads/master"},"6adb26135391e94a7a51c0f4af43a530652493bd":{"kind":"REWORK","_number":3,"created":"2021-03-09 11:09:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/70/773070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/773070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/773070/3"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:28:31.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-09 11:09:01.000000000","tz":0},"subject":"Release networking-ovn for stable/train","message":"Release networking-ovn for stable/train\n\nThis release picks up new commits to networking-ovn since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id029fed75a72858eec9d983bb015007c144c96b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6adb26135391e94a7a51c0f4af43a530652493bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6adb26135391e94a7a51c0f4af43a530652493bd"}]},"branch":"refs/heads/master"},"5e456493b254f67885225854309916643c0e467a":{"kind":"REWORK","_number":4,"created":"2021-03-09 20:00:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/773070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/773070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/773070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/773070/4"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:28:31.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-09 19:59:59.000000000","tz":-300},"subject":"Release networking-ovn for stable/train","message":"Release networking-ovn for stable/train\n\nThis release picks up new commits to networking-ovn since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id029fed75a72858eec9d983bb015007c144c96b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e456493b254f67885225854309916643c0e467a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e456493b254f67885225854309916643c0e467a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
