)]}'
{"id":"openstack%2Freleases~773074","triplet_id":"openstack%2Freleases~master~I1f8063b3151426183e349a8721d2bce89db37be8","project":"openstack/releases","branch":"master","topic":"nova-train-20.5.0","hashtags":[],"change_id":"I1f8063b3151426183e349a8721d2bce89db37be8","subject":"nova: Release Train 20.5.0","status":"MERGED","created":"2021-01-29 14:36:36.000000000","updated":"2021-02-04 10:36:52.000000000","submitted":"2021-02-04 10:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"773074-nova-train-20.5.0","meta_rev_id":"5a35f469005a4398212d141565c135877edb1270","_number":773074,"virtual_id_number":773074,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 10:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-02-04 10:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2021-02-02 12:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-01 16:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-02-04 10:23:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-02-01 16:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 14:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 15:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 15:24:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-01-29 15:24:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-02-01 13:18:06.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-02-02 12:23:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-04 10:23:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"1b43a70de3faef576dc756870d462d118be2c537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-29 14:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ee07c368a1cdf448173215edfb4cdceb49f7227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 15:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5eb6c0bef324e8687e6f39844b0bc7c : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4c721770a4140a985931f73cdd3b1d4 : SUCCESS in 41m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2c8c91d39e4f489c9d3ae9619c65c0eb : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e355c5abc180823a845fd1db7d6198fd7109e3a7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-01 09:54:51.000000000","message":"Patch Set 1: Code-Review-1\n\nTalked to Elod and we agreed to release stable/victoria and stable/ussuri together with stable/train. So the -1 is procedural until we get those releases lined up","accounts_in_message":[],"_revision_number":1},{"id":"f799d25e4fae2b56d7f7442daa9055131fb7836f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-01 10:21:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"38e428d89392775ae27ecc0aff9e768db588eb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 11:03:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7121fdbf6c4f64a319a69023735d42 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6b9b29fadc84cdc86b2dfcf0a37fd83 : SUCCESS in 39m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/34ee1590c8b44d8cac6318f3bee01a83 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fbe2fd1a4bb4afd55d48806cdbb4da4a9cc63db5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-02-01 13:18:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bd55d2db33c0e182fdfa7c3692ecc50e4687d7db","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-01 13:52:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee22b2168aadd665e0a904225a5af46daf80389d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-01 14:05:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a14616378472ab04fbfcc8a5719be3203d4b651","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-01 14:08:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40f7feb3053a71579c8de191eec21618a003412c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 14:17:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08f4f71e12c4ae0a955f1ad38558f58 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/883e4ac8b0354c8bbbc0b3c8947b1e42 : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/88a0eb7a030145e593a4f5b21f705ee6 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8872ea4064b5c17bf91a794a817d5a083a06b011","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-01 16:03:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea37770b0e125a03916fd68e133a4f82bf1a0b94","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:05:32.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dbabfca56ed35d5b4175b38b33573109eab3d21a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-02 12:23:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f2e458eb0a40e9ffcea989e59056e35cc6f10601","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-04 10:23:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"794da7a451fc6291473141ae2db47c5deeaf0978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:24:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01ab015abb9745be8d1f4a91d0d309047bee8302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:36:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902b14dd1e6a4095904ab4e1a70e1bb8 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da63903161b44278ad82398b35703961 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c5f6d5ac2af4195a27de824b220fa79 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"5a35f469005a4398212d141565c135877edb1270","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:36:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"05dc1f0bae87f899ea1d9f03b875c1f9874d1def","revisions":{"a7ad8e63f3e79491f5a9cb5834e5e81698c9ff20":{"kind":"REWORK","_number":1,"created":"2021-01-29 14:36:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/773074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/773074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/773074/1"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:36:30.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:36:30.000000000","tz":60},"subject":"Release nova for stable/train","message":"Release nova for stable/train\n\nThis release picks up new commits to nova since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1f8063b3151426183e349a8721d2bce89db37be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7ad8e63f3e79491f5a9cb5834e5e81698c9ff20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7ad8e63f3e79491f5a9cb5834e5e81698c9ff20"}]},"branch":"refs/heads/master"},"188753c939ed61120e8ec972c6dbc72c6d85441d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-01 10:21:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/773074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/773074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/773074/2"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:36:30.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-01 10:17:50.000000000","tz":60},"subject":"nova: Release Train 20.4.2","message":"nova: Release Train 20.4.2\n\n$ git log --oneline --no-merges 20.4.1..HEAD\n8378785f99 Update pci stat pools based on PCI device changes\nb2037fc4e3 [stable-only] Cap bandit to 1.6.2 and raise hacking, flake8 and stestr\n4653245ddc Validate id as integer for os-aggregates\n0c0c5b12cf docs: Clarify configuration steps for PF devices\n9223613ad3 docs: Change order of PCI configuration steps\n74b2af473c docs: Rework the PCI passthrough guides\nca2fd8098a Change default num_retries for glance to 3\n286d7cfc5c add [libvirt]/max_queues config option\n\nChange-Id: I1f8063b3151426183e349a8721d2bce89db37be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/188753c939ed61120e8ec972c6dbc72c6d85441d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/188753c939ed61120e8ec972c6dbc72c6d85441d"}]},"branch":"refs/heads/master"},"05dc1f0bae87f899ea1d9f03b875c1f9874d1def":{"kind":"REWORK","_number":3,"created":"2021-02-01 14:05:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/773074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/773074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/773074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/773074/3"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 14:36:30.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-01 14:02:28.000000000","tz":60},"subject":"nova: Release Train 20.5.0","message":"nova: Release Train 20.5.0\n\nMinor version bump is needed because of the new config option in ptach\nI9abe783a9a9443c799e7c74a57cc30835f679a01\n(286d7cfc5c add [libvirt]/max_queues config option).\n\n$ git log --oneline --no-merges 20.4.1..HEAD\n8378785f99 Update pci stat pools based on PCI device changes\nb2037fc4e3 [stable-only] Cap bandit to 1.6.2 and raise hacking, flake8 and stestr\n4653245ddc Validate id as integer for os-aggregates\n0c0c5b12cf docs: Clarify configuration steps for PF devices\n9223613ad3 docs: Change order of PCI configuration steps\n74b2af473c docs: Rework the PCI passthrough guides\nca2fd8098a Change default num_retries for glance to 3\n286d7cfc5c add [libvirt]/max_queues config option\n\nChange-Id: I1f8063b3151426183e349a8721d2bce89db37be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05dc1f0bae87f899ea1d9f03b875c1f9874d1def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/05dc1f0bae87f899ea1d9f03b875c1f9874d1def"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
