)]}'
{"id":"openstack%2Freleases~773133","triplet_id":"openstack%2Freleases~master~Iee05fa590699c2aa67907d9eefe1b53a9a57b4f5","project":"openstack/releases","branch":"master","topic":"train-stable","hashtags":[],"change_id":"Iee05fa590699c2aa67907d9eefe1b53a9a57b4f5","subject":"Release puppet-mistral for stable/train","status":"ABANDONED","created":"2021-01-29 16:12:31.000000000","updated":"2021-02-16 11:50:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c437e145a8d9292f868352db140e5abf66d2928","_number":773133,"virtual_id_number":773133,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-29 17:17:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 16:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 16:20:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-01-29 17:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91e0c5636997ae5c54ea0386ab1c617544f49d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-29 16:12:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d381db1e0ecc73b2eab5fd3a5d4b7f027d7d2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 17:17:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02f353b3511468090360df109ad6a30 : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8681fa4e479a4e6f87b81c00858d5c57 : FAILURE in 4m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/01c44dcb99a3465f8df1172de68bdacb : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4c437e145a8d9292f868352db140e5abf66d2928","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-16 11:50:25.000000000","message":"Abandoned\n\nNo response from team, and this was just proposed as a way to help make sure stable commits were actually being released. So abandoning this patch, but feel free to restore at any point, or propose a new release request, when ready for a stable release.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e1de1d2c1725f01337548b1691df359af6cb47d","revisions":{"1e1de1d2c1725f01337548b1691df359af6cb47d":{"kind":"REWORK","_number":1,"created":"2021-01-29 16:12:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/33/773133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/773133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/773133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/773133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/773133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/773133/1"}}},"commit":{"parents":[{"commit":"81dbde72df01627260cdebc5af6b03cc0dcc36fa","subject":"Merge \"Octavia: EOL Stein branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81dbde72df01627260cdebc5af6b03cc0dcc36fa"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 16:12:25.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-29 16:12:25.000000000","tz":60},"subject":"Release puppet-mistral for stable/train","message":"Release puppet-mistral for stable/train\n\nThis release picks up new commits to puppet-mistral since\nthe last release from stable/train.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a  stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iee05fa590699c2aa67907d9eefe1b53a9a57b4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e1de1d2c1725f01337548b1691df359af6cb47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e1de1d2c1725f01337548b1691df359af6cb47d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
