)]}'
{"id":"openstack%2Freleases~774071","triplet_id":"openstack%2Freleases~master~I8f26fb5655ef1cd280b243cadbb7b49d596e7d51","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I8f26fb5655ef1cd280b243cadbb7b49d596e7d51","subject":"Optional list of changes in commit message for auto release","status":"MERGED","created":"2021-02-04 11:36:11.000000000","updated":"2021-02-25 15:52:18.000000000","submitted":"2021-02-25 15:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774071","meta_rev_id":"2577353a2d4f7d6e443decd3af9e02b4e2136ad4","_number":774071,"virtual_id_number":774071,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 15:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-25 15:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-02-25 15:10:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-25 15:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 11:39:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-02-04 11:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 14:25:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-02-18 10:04:12.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-02-25 10:25:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-02-25 15:10:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-25 15:32:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"834534202b1c876a5a8956d57646679cd61f737e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-04 11:36:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41a15a596e538088d1082d0c1116e6414a51b003","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 11:39:28.000000000","message":"Patch Set 1:\n\nI think you should also provide flag to allow to disable that, or even a commit message placeholder as already did with \u0027PROJECT\u0027","accounts_in_message":[],"_revision_number":1},{"id":"ecef980883c332ebac1f57f76ffffde3ce1568a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff34a0fdcb5446fb5cd58e3f19a0b3e : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c3f6d20895844a78da216e929781927 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad98ae40df954493893775410d5cb4bd : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e7c9ac95e84450a204095f52189790 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63d9c83ae0ee4a38ba5829e8c6a7ae43 : SUCCESS in 4m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e9f0d7fdc01642b9a57d76f335f7b1c3 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08f1a41966c147eba58951d6fcb89517 : SUCCESS in 4m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/897a7b8804814213973cb520e566d89c : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6f67c3ad9400f9f76e04971c6a584bd40c9b960f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 14:25:45.000000000","message":"Patch Set 1:\n\nDo we really want all of this in the commit message? That\u0027s kind of what our list-changes job is for.","accounts_in_message":[],"_revision_number":1},{"id":"e19ed722d7214c7b143c09f1fabbf83e8f86ba8e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-18 10:04:12.000000000","message":"Patch Set 1:\n\n@Sean: Accessing and parsing the list-changes output takes a lot of time, and recently we had a ton of stable release requests. Putting the shortlist of changes in the commit message allows to quickly scan for potential issues (which would trigger a full inspection of the list-changes output). Nova\u0027s stable team provides those types of messages and I find them useful.","accounts_in_message":[],"_revision_number":1},{"id":"a41a189b30e7ab38651ecabd834d36ed726eb2de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-24 19:38:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d28432057fcab727090303ec402bb298fc9a162","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-24 19:49:13.000000000","message":"Patch Set 2:\n\nSorry for not coming back earlier to this patch.\n\nBasically this is some extra (duplicate) information in the commit message, indeed. Maybe this is good in case of the mass-generated, before-EM stable patches (like: [1]). See an example release patch from nova [2].\n\n[1] https://review.opendev.org/q/topic:train-stable\n[2] https://review.opendev.org/773344","accounts_in_message":[],"_revision_number":2},{"id":"4c0314f1447272afddb14ce69fb64744090417d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 19:51:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd6bfb0e968417eba2f0fedf6efa37e : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33460a9c5614fb58e8960bbd4c367c2 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c702ed3756344459a77a22b29af6d60 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196a2d4738fd47ed9827d97da92ab913 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/727de642ae4848a9b09fd8b573c3484e : SUCCESS in 5m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/05bca2dc2b514ce5b84dce76ec674fce : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff825af7afe04015a71b63b37e2c2aff : SUCCESS in 7m 17s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0c94396e9f634d34a1d25829f2423a0b : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2fa155682a94023d8444858a1930dc007124ea90","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-24 20:41:32.000000000","message":"Patch Set 2:\n\n\u003e Basically this is some extra (duplicate) information in the commit message, indeed. Maybe this is good in case of the mass-generated, before-EM stable patches (like: [1]). See an example release patch from nova [2].\n\u003e \n\nPersonally, I\u0027ve tried to discourage nova from doing that with \"Hey, you don\u0027t really need to do that\" messages in the past. I don\u0027t think we need to clutter up our commit messages with information that is easily attained by running a git diff or other means.\n\nBut that is just my personal opinion. Really no harm other than git data bloat.","accounts_in_message":[],"_revision_number":2},{"id":"5a22cdc92b5fae41abbbb3c450a1e834bcddbbd8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 08:40:32.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Basically this is some extra (duplicate) information in the commit message, indeed. Maybe this is good in case of the mass-generated, before-EM stable patches (like: [1]). See an example release patch from nova [2].\n\u003e \u003e \n\u003e \n\u003e Personally, I\u0027ve tried to discourage nova from doing that with \"Hey, you don\u0027t really need to do that\" messages in the past. I don\u0027t think we need to clutter up our commit messages with information that is easily attained by running a git diff or other means.\n\u003e \n\u003e But that is just my personal opinion. Really no harm other than git data bloat.\n\nHmmm. I did not know that story, I just continued the nova team\u0027s way of formatting their release patches\u0027 commit messages. I found it convenient, but you are right that one can easily check it with git, and of course the information is already there in list-changes-results.log as well.\n\nSo I accept your view and now I\u0027m unsure whether we really need this... o:)","accounts_in_message":[],"_revision_number":2},{"id":"26bf9a8dd1465647177374db24aa51871519d359","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-25 10:21:21.000000000","message":"Patch Set 2:\n\nPersonally I think it\u0027s valuable and can save up reviewer time (which is why we create all this automation after all). Of course the data is already available somewhere (otherwise the automation would not be able to find it), the question is how much human time do you want to spend gathering it.\n\nIf I have the version number and the list of changes in the commit message, I\u0027m able to short-circuit a number of changes and not check list-changes (which is very time-consuming as you need to scroll down to get to the interesting parts of the log)\n\nFor example on minor bumps a quick check on the commit messages is enough to validate the change. Some patch bumps on CI changes can also be quickly validated. A good example is the recent dump of train stable releases, where most of them were CI-only things that could quickly be reviewed. Having to open and parse list-changes on dozens of changes just to check if the changes was CI-only was not a lot of fun.\n\nBloating the releases git repository with slightly longer commit messages sounds like a pretty small price to pay.","accounts_in_message":[],"_revision_number":2},{"id":"46b5ab5bee5ce9fdbc74a872b2458014abf03932","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-25 10:25:19.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM\n\nI like to have version numbers in the commit message title too (future improvement)","accounts_in_message":[],"_revision_number":2},{"id":"c9714c527cd6d236d405dfff40d57972e9547bd1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 10:45:10.000000000","message":"Patch Set 2:\n\n\u003e A good example is the recent dump of train stable releases, where most of them were CI-only things that could quickly be reviewed. Having to open and parse list-changes on dozens of changes just to check if the changes was CI-only was not a lot of fun.\n\nWell, actually that was my fault o:) I\u0027ve filtered out lots of CI-only changes, but where I was not so confident I let the project teams to review them. Sorry for that :S\n\nAnyway, maybe it is a good trade, to mainly use this in case of mass-generated release patches.","accounts_in_message":[],"_revision_number":2},{"id":"493d1792b1f76da9424a89e83394fe9d6144a8ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 10:51:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd51b3bcbc6c350967e4fc5bc314fb02f331ec39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-25 10:54:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dcb7f989db7e39f2a135ad154a1fc55d6e10a098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:07:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafdbc318b5b40a1adc30d7e734e2c3d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e4c265ddca431b9edbaeef15f6e2d0 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5368deb7b270469a9aaae65880d98e9a : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f11165195c4ba9b675ef090242d99e : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d36511c1f924f4eb1cc7599e5d5200e : SUCCESS in 4m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/80cd66b8d4ce4ce48176886476575bb0 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a607ef70cd14bbfa575109640011f41 : SUCCESS in 4m 33s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4f70e5c8bef0423c8a615af3e5f7b307 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8554885639d6e03d66ed45b73ffe2ac4ea53cc04","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-25 15:10:38.000000000","message":"Patch Set 4: Code-Review+2\n\nI agree with Sean for isolated patches proposed by PTL or liaison we don\u0027t really need to put the history in the commit message.\n\nHowever in this case (batches of patches) I think that it\u0027s useful for us, especially when douzens of patches are proposed automatically.","accounts_in_message":[],"_revision_number":4},{"id":"3a5c285a83003890a1cffd126c475a164755dcc4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-25 15:32:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nYeah, makes sense. At least it is an optional behavior.","accounts_in_message":[],"_revision_number":4},{"id":"2980115c19b014b53ee7cc4412e6a7e3040175fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 15:33:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2ee200e9fc5dbfcd12ede68b3fc85be56e0f6227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 15:52:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c27cf6de364570a3b8c0335e50cee5 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bd1de3330024f09823b187ff76b1a6a : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1580f054f50c4c919bb63d0bd0a0ae3a : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05fe4738c6de4c45a81d2d7d8ea0e382 : SUCCESS in 9m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06697c3050244182810192392677e47f : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56e674cf65424f4dadec718bd1e7cb9b : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2577353a2d4f7d6e443decd3af9e02b4e2136ad4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 15:52:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b293ca4bcbf95b9a85ecf9a4c786371ec9aa3e71","revisions":{"b70baf372e4dd1a5e0b77f3ffc70b3ea73dc700a":{"kind":"REWORK","_number":1,"created":"2021-02-04 11:36:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/774071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/774071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/774071/1"}}},"commit":{"parents":[{"commit":"c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08","subject":"Merge \"Release kuryr-kubernetes for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-04 11:35:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-04 11:35:12.000000000","tz":60},"subject":"WIP: Append change list at the end of the commit message","message":"WIP: Append change list at the end of the commit message\n\nChange-Id: I8f26fb5655ef1cd280b243cadbb7b49d596e7d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b70baf372e4dd1a5e0b77f3ffc70b3ea73dc700a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b70baf372e4dd1a5e0b77f3ffc70b3ea73dc700a"}]},"branch":"refs/heads/master"},"fd2a639424ae4f37d7c55876ce11aa2e260d49f5":{"kind":"REWORK","_number":2,"created":"2021-02-24 19:38:38.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/774071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/774071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/774071/2"}}},"commit":{"parents":[{"commit":"c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08","subject":"Merge \"Release kuryr-kubernetes for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-04 11:35:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-24 19:38:17.000000000","tz":60},"subject":"Optional list of changes in commit message for auto release","message":"Optional list of changes in commit message for auto release\n\nWhen generating release patches with process_auto_releases.sh it could\nbe useful to have the list of changes in the commit message of the\nrelease patch. This patch adds an extra question for that.\n\nChange-Id: I8f26fb5655ef1cd280b243cadbb7b49d596e7d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd2a639424ae4f37d7c55876ce11aa2e260d49f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd2a639424ae4f37d7c55876ce11aa2e260d49f5"}]},"branch":"refs/heads/master"},"335c3f91355112467c1aaa1b0e32e7638a67107f":{"kind":"REWORK","_number":3,"created":"2021-02-25 10:51:50.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/774071/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/774071/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/774071/3"}}},"commit":{"parents":[{"commit":"c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08","subject":"Merge \"Release kuryr-kubernetes for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-04 11:35:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-25 10:51:14.000000000","tz":60},"subject":"Optional list of changes in commit message for auto release","message":"Optional list of changes in commit message for auto release\n\nWhen generating release patches with process_auto_releases.sh it could\nbe useful to have the list of changes in the commit message of the\nrelease patch. This patch adds an extra question for that.\n\nChange-Id: I8f26fb5655ef1cd280b243cadbb7b49d596e7d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/335c3f91355112467c1aaa1b0e32e7638a67107f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/335c3f91355112467c1aaa1b0e32e7638a67107f"}]},"branch":"refs/heads/master"},"b293ca4bcbf95b9a85ecf9a4c786371ec9aa3e71":{"kind":"REWORK","_number":4,"created":"2021-02-25 10:54:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/774071/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/774071/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/774071/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/774071/4"}}},"commit":{"parents":[{"commit":"c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08","subject":"Merge \"Release kuryr-kubernetes for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e36ebc9ea6f4eab5eaa925f6037cba19cc2a08"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-04 11:35:12.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-02-25 10:53:52.000000000","tz":60},"subject":"Optional list of changes in commit message for auto release","message":"Optional list of changes in commit message for auto release\n\nWhen generating release patches with process_auto_releases.sh it could\nbe useful to have the list of changes in the commit message of the\nrelease patch. This patch adds an extra question for that.\n\nChange-Id: I8f26fb5655ef1cd280b243cadbb7b49d596e7d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b293ca4bcbf95b9a85ecf9a4c786371ec9aa3e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b293ca4bcbf95b9a85ecf9a4c786371ec9aa3e71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
