)]}'
{"id":"openstack%2Freleases~774244","triplet_id":"openstack%2Freleases~master~I3523b079b457b8f94780bd7c8f6e616f81f6eb40","project":"openstack/releases","branch":"master","topic":"I3523b079b457b8f94780bd7c8f6e616f81f6eb40","hashtags":[],"change_id":"I3523b079b457b8f94780bd7c8f6e616f81f6eb40","subject":"Tag stable/rocky tripleo repos as EOL","status":"MERGED","created":"2021-02-05 14:32:02.000000000","updated":"2021-03-05 17:40:50.000000000","submitted":"2021-03-05 17:40:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"774244-I3523b079b457b8f94780bd7c8f6e616f81f6eb40","meta_rev_id":"db81eefde815409c37431f903f153073730ae9eb","_number":774244,"virtual_id_number":774244,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 17:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-03-05 16:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-03-05 17:32:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":2,"date":"2021-03-05 17:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":1,"date":"2021-03-05 17:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-05 16:47:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 14:33:05.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-02-05 14:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 15:16:58.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-02-05 15:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-02-08 14:32:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-02-09 12:04:12.000000000","updated_by":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"reviewer":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"state":"REVIEWER"},{"updated":"2021-02-09 13:36:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-02-10 09:46:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-22 14:39:17.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-02-22 14:39:44.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-02-23 15:47:13.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-03-01 12:24:30.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-01 17:28:08.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REMOVED"},{"updated":"2021-03-02 14:37:33.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"ba99b8dcb622a9297af072b613c4158d13722d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-05 14:32:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da844da201e6ef738ef70d1d1052595e60a1374d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-05 14:32:48.000000000","message":"Patch Set 1: Workflow-1\n\nwf-1 until we are sure sure ;) waiting for last chance comments via the mailing list","accounts_in_message":[],"_revision_number":1},{"id":"36255c70da582f9d0338f4f7e8c865d625a1c645","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:33:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d99dd6a4edcc2b3b70e9b16d0796e42ad29cbebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:40:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bac9af3941a4c418e5772e4de75ab46 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d77e1e834e1b46b49ee8d00681237d7d : SUCCESS in 5m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2e939b19eecc4a689f6326899ad7246a : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"79cbd77a08f97b761fabdf3e4ff4c938fa62a965","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-05 15:16:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1774207499011b001499209743bace5e6840764c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-08 14:32:44.000000000","message":"Patch Set 1: Code-Review-1\n\nStrangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.","accounts_in_message":[],"_revision_number":1},{"id":"c7c04c9704c71a1ae188cb4bf08491673b059e80","author":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"date":"2021-02-09 12:04:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd97a059af9bbe7d29c899b6f9586cc81b7e3f8d","author":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"date":"2021-02-09 12:08:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Strangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.\n\nTripleo still heavily uses queens as part of a \"multiple version upgrade\", ie upgrading from Queens to Train. So it Tripleo\u0027s context it makes sense to EOL Rocky without EOL Queens.","accounts_in_message":[],"_revision_number":1},{"id":"8ea15a64da7d70a7811fb3a9a44475cf3e9a67c4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-09 13:17:12.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Strangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.\n\u003e \n\u003e Tripleo still heavily uses queens as part of a \"multiple version upgrade\", ie upgrading from Queens to Train. So it Tripleo\u0027s context it makes sense to EOL Rocky without EOL Queens.\n\nThen I don\u0027t really understand why do you need to EOL Rocky, as it is a \u0027middle\u0027 branch between Train and Queens.\nProjects which are following the stable policy [1] cannot really \u0027skip\u0027 any branch. Neither is it possible to EOL a \u0027middle\u0027 branch for them according to the current policy.\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes","accounts_in_message":[],"_revision_number":1},{"id":"77149565cb559b8d269fae98dc9c749243fd4b2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-09 13:36:02.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e Strangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.\n\u003e \u003e \n\u003e \u003e Tripleo still heavily uses queens as part of a \"multiple version upgrade\", ie upgrading from Queens to Train. So it Tripleo\u0027s context it makes sense to EOL Rocky without EOL Queens.\n\u003e \n\u003e Then I don\u0027t really understand why do you need to EOL Rocky, as it is a \u0027middle\u0027 branch between Train and Queens.\n\u003e Projects which are following the stable policy [1] cannot really \u0027skip\u0027 any branch. Neither is it possible to EOL a \u0027middle\u0027 branch for them according to the current policy.\n\u003e \n\u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes\n\nGood point.\n\nUnfortunatelly in our process moving to EOL will trigger a branch deletion... and I agree with Elod skip a middle branch during a backport can lead to inconsistences, and AFAIK this is not something allowed by our policy.\n\nHowever tripleo doesn\u0027t follow the stable policy https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n\nOn the other side AFAIK tripleo is only used by redhat, isn\u0027t?\nI think that yes, but if this is not the case then this point should matter too.\n\nrocky is RHOSP14, queens is RHOSP13, RHOSP14 is EOL, RHOSP13 is still maintained on downstream/redhat side, so from a redhat point of view this isn\u0027t an issue, but first we should ensure that this point is ok (tripleo co-users).\n\nPlease can you confirm that point?","accounts_in_message":[],"_revision_number":1},{"id":"b299a13d9e1817bbff013f881f88380e702e6246","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-09 14:22:23.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Strangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.\n\u003e \u003e \u003e \n\n\nyeah, as pointed out by others here, Queens is special for TripleO as one of the fast forward upgrade starting points (newton--\u003equeens--\u003etrain--\u003ewallaby)\n\nOur stable/queens branches are still seeing some activity but we expect that to start to diminish over the next few months. So we are a few months out for moving queens to eol. \n\nHowever rocky has seen very little activity over the last year or so, which is why we are proposing that to eol (you can see more info about the activy at http://paste.openstack.org/raw/800464/ via http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html where i first proposed this a couple months ago\n\n\n\n\u003e \u003e \u003e Tripleo still heavily uses queens as part of a \"multiple version upgrade\", ie upgrading from Queens to Train. So it Tripleo\u0027s context it makes sense to EOL Rocky without EOL Queens.\n\u003e \u003e \n\u003e \u003e Then I don\u0027t really understand why do you need to EOL Rocky, as it is a \u0027middle\u0027 branch between Train and Queens.\n\u003e \u003e Projects which are following the stable policy [1] cannot really \u0027skip\u0027 any branch. Neither is it possible to EOL a \u0027middle\u0027 branch for them according to the current policy.\n\u003e \u003e \n\u003e \u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes\n\u003e \n\u003e Good point.\n\u003e \n\u003e Unfortunatelly in our process moving to EOL will trigger a branch deletion... and I agree with Elod skip a middle branch during a backport can lead to inconsistences, and AFAIK this is not something allowed by our policy.\n\u003e \n\u003e However tripleo doesn\u0027t follow the stable policy https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n\u003e \n\u003e On the other side AFAIK tripleo is only used by redhat, isn\u0027t?\n\u003e I think that yes, but if this is not the case then this point should matter too.\n\u003e \n\u003e rocky is RHOSP14, queens is RHOSP13, RHOSP14 is EOL, RHOSP13 is still maintained on downstream/redhat side, so from a redhat point of view this isn\u0027t an issue, but first we should ensure that this point is ok (tripleo co-users).\n\u003e \n\u003e Please can you confirm that point?\n\n\nthis is precisely what i have been asking for on the mailing list first there http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html then most recently here http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html i.e. does anyone object to stable/rocky going eol? I haven\u0027t heard any objection yet","accounts_in_message":[],"_revision_number":1},{"id":"cdc51f579faf94219ea1a45cc904a9e2acf125ad","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-09 15:44:28.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e \u003e Strangely it seems Queens is not EOL yet. To move Rocky to EOL first Queens needs to be EOL\u0027d.\n\u003e \u003e \u003e \u003e \n\u003e \n\u003e \n\u003e yeah, as pointed out by others here, Queens is special for TripleO as one of the fast forward upgrade starting points (newton--\u003equeens--\u003etrain--\u003ewallaby)\n\u003e \n\u003e Our stable/queens branches are still seeing some activity but we expect that to start to diminish over the next few months. So we are a few months out for moving queens to eol. \n\u003e \n\u003e However rocky has seen very little activity over the last year or so, which is why we are proposing that to eol (you can see more info about the activy at http://paste.openstack.org/raw/800464/ via http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html where i first proposed this a couple months ago\n\u003e \n\u003e \n\u003e \n\u003e \u003e \u003e \u003e Tripleo still heavily uses queens as part of a \"multiple version upgrade\", ie upgrading from Queens to Train. So it Tripleo\u0027s context it makes sense to EOL Rocky without EOL Queens.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Then I don\u0027t really understand why do you need to EOL Rocky, as it is a \u0027middle\u0027 branch between Train and Queens.\n\u003e \u003e \u003e Projects which are following the stable policy [1] cannot really \u0027skip\u0027 any branch. Neither is it possible to EOL a \u0027middle\u0027 branch for them according to the current policy.\n\u003e \u003e \u003e \n\u003e \u003e \u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes\n\u003e \u003e \n\u003e \u003e Good point.\n\u003e \u003e \n\u003e \u003e Unfortunatelly in our process moving to EOL will trigger a branch deletion... and I agree with Elod skip a middle branch during a backport can lead to inconsistences, and AFAIK this is not something allowed by our policy.\n\u003e \u003e \n\u003e \u003e However tripleo doesn\u0027t follow the stable policy https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html\n\u003e \u003e \n\u003e \u003e On the other side AFAIK tripleo is only used by redhat, isn\u0027t?\n\u003e \u003e I think that yes, but if this is not the case then this point should matter too.\n\u003e \u003e \n\u003e \u003e rocky is RHOSP14, queens is RHOSP13, RHOSP14 is EOL, RHOSP13 is still maintained on downstream/redhat side, so from a redhat point of view this isn\u0027t an issue, but first we should ensure that this point is ok (tripleo co-users).\n\u003e \u003e \n\u003e \u003e Please can you confirm that point?\n\u003e \n\u003e \n\u003e this is precisely what i have been asking for on the mailing list first there http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html then most recently here http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html i.e. does anyone object to stable/rocky going eol? I haven\u0027t heard any objection yet\n\nIn the original mail Queens was not mentioned (\"This will allow us to focus our efforts on keeping the check and gate queues green and continue to deliver weekly promotions for the more recent and active stable/* branches train ussuri victoria and master.\")\n\nAnyway, as a stable core, I still think this is a bit unfortunate. I really would suggest to close Queens first - or wait with Rocky EOL\u0027ing until Queens get there, too (especially as keeping queens alive would not mean generally such extra effort to keep rocky alive, too).\n\nNevertheless, I understand that TripleO does not follow stable-policy, so I remove my -1.","accounts_in_message":[],"_revision_number":1},{"id":"75b2b4bf4b6b2c8bff1d5e13c59332fd7b40c528","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-10 09:46:17.000000000","message":"Patch Set 1: Code-Review+1\n\nFrom a downstream perspective RHOSP13 will be EOL in June 2021 so I don\u0027t expect lot of activity on Queens upstream too. I don\u0027t expect lot backports on something nearly dead.\n\nFrom a stable branch perspective as already said tripleo doesn\u0027t subscribe to this policy, so we can\u0027t block this for this reason. Thanks Elod for deleting your vote.\n\nFrom a QA/testing perspective I think that these changes could allow the tripleo team to reduce their workload.\n\nFrom a \"broad usage\" point of view, Mario you\u0027re right nobody complained about this.\n\nFor all these reasons I think we can go ahead.\n\n@Mario I only leave a +1 until you remove your -W, I\u0027ll update my vote when you\u0027ll unlock the patch.","accounts_in_message":[],"_revision_number":1},{"id":"0c50cbe9e4626de5991be5d3bd58e5d344e10efe","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-10 10:31:46.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e From a downstream perspective RHOSP13 will be EOL in June 2021 so I don\u0027t expect lot of activity on Queens upstream too. I don\u0027t expect lot backports on something nearly dead.\n\u003e \n\u003e From a stable branch perspective as already said tripleo doesn\u0027t subscribe to this policy, so we can\u0027t block this for this reason. Thanks Elod for deleting your vote.\n\u003e \n\u003e From a QA/testing perspective I think that these changes could allow the tripleo team to reduce their workload.\n\u003e \n\u003e From a \"broad usage\" point of view, Mario you\u0027re right nobody complained about this.\n\u003e \n\u003e For all these reasons I think we can go ahead.\n\u003e \n\u003e @Mario I only leave a +1 until you remove your -W, I\u0027ll update my vote when you\u0027ll unlock the patch.\n\n\nthanks Herve and Elod for your careful consideration of these reviews.\n\n@Herve I will leave my -1 in place for a few more days, blocking on the upstream conversation, if any @ [1] so no-one can possibly say there wasn\u0027t enough notification. Besides the upstream list I have also brought it up in our irc meeting on more than one occasion recently.\n \n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html","accounts_in_message":[],"_revision_number":1},{"id":"fd42b5a72e6766fbb1fa384f9221bf02a64d7ffe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-11 15:13:43.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e From a downstream perspective RHOSP13 will be EOL in June 2021 so I don\u0027t expect lot of activity on Queens upstream too. I don\u0027t expect lot backports on something nearly dead.\n\u003e \u003e \n\u003e \u003e From a stable branch perspective as already said tripleo doesn\u0027t subscribe to this policy, so we can\u0027t block this for this reason. Thanks Elod for deleting your vote.\n\u003e \u003e \n\u003e \u003e From a QA/testing perspective I think that these changes could allow the tripleo team to reduce their workload.\n\u003e \u003e \n\u003e \u003e From a \"broad usage\" point of view, Mario you\u0027re right nobody complained about this.\n\u003e \u003e \n\u003e \u003e For all these reasons I think we can go ahead.\n\u003e \u003e \n\u003e \u003e @Mario I only leave a +1 until you remove your -W, I\u0027ll update my vote when you\u0027ll unlock the patch.\n\u003e \n\u003e \n\u003e thanks Herve and Elod for your careful consideration of these reviews.\n\u003e \n\u003e @Herve I will leave my -1 in place for a few more days, blocking on the upstream conversation, if any @ [1] so no-one can possibly say there wasn\u0027t enough notification. Besides the upstream list I have also brought it up in our irc meeting on more than one occasion recently.\n\u003e  \n\u003e \n\u003e [1] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html\n\nNo problem make sense. Thanks for the details :)","accounts_in_message":[],"_revision_number":1},{"id":"85745b9a8426a914d0de7c04474420bc4c46c05a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-12 15:34:13.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)\n\nOK it has been a week since I posted http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html so I think it is appropriate that I remove my blocking WF -1\n\n\nAdded inline note/explanation about a peculiarity with tripleo-ipsec and a missing tag vs validations.\n\nthank you for reviewing","accounts_in_message":[],"_revision_number":1},{"id":"c195494893db4aa085eec4cfa8d973b5fb35b257","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-12 15:34:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce8a4066a7d7477b0af08b51ed36bdcd7765944a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 10:58:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8bf7300209a4f1710a259a91e3d852c16af7d64","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-22 14:39:17.000000000","message":"Assignee added: Thierry Carrez \u003cthierry@openstack.org\u003e","accounts_in_message":[],"_revision_number":1},{"id":"64dcff892bcba4df2e17b0331968751393edd4ae","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-22 14:39:29.000000000","message":"Assignee deleted: Thierry Carrez \u003cthierry@openstack.org\u003e","accounts_in_message":[],"_revision_number":1},{"id":"bcca718467085777a69272fca99728606e9c394a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-24 13:11:30.000000000","message":"Patch Set 1:\n\nbump\n\ngrateful for your reviews please anyone know who else we can ask to look at this","accounts_in_message":[],"_revision_number":1},{"id":"1a704d26c8892f150f273340db343300500fb4df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 17:35:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI wanted to approve this, but I\u0027ve found some small issues... sorry o:)\nCould you please check them?\n\n- I see that openstack/instack and openstack/tripleo-ui is missing from the EOL list [1]. Is that intentional? Do you want to keep those open?\n- tripleo-heat-template\u0027s proposed EOL tag is not on the last commit\n\n[1] https://zuul.opendev.org/t/openstack/build/2e939b19eecc4a689f6326899ad7246a/log/tox/list-changes-results.log#26-42","accounts_in_message":[],"_revision_number":1},{"id":"8e4eb481d9553703d502e9e50171c64def52bd17","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 06:49:22.000000000","message":"Patch Set 1: -Code-Review\n\nDropping my +2 to highlight Elod\u0027s comment","accounts_in_message":[],"_revision_number":1},{"id":"e328a0c2a5df22802d97698fe7a826583f3fe6fb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-01 09:29:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nthank you Elod and Herve for taking the time to give such careful reviews and make sure we don\u0027t do anything unwanted ;)","accounts_in_message":[],"_revision_number":1},{"id":"18324347b879828631706d624ebfcacc24a425cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-01 09:41:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b923fc3e8aaaaf13030c7c164b199a14e020051","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:43:43.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"631c606585fa8c8f4de488612c8462da94de47f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:09:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9fba2307b248fcbc905a017a197823 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f1c4ba03a54433e9fe4abf949631fe8 : SUCCESS in 5m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/274b64af2e1149e086b8ee5690790da7 : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":2},{"id":"daef5dd1e9c13a4d1dc35f150d849868f8b8565b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-01 10:22:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdaf3479a8695381650df74f6b8196f6cb8a653c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-01 12:24:30.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nOK by me.","accounts_in_message":[],"_revision_number":2},{"id":"4f8301e2eb6eaaa8e1e322be0503f3768b65cb44","author":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"date":"2021-03-01 17:04:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"03cb7346b4b3357bab20eeb11ac1211bb5792c4d","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-03-01 17:28:08.000000000","message":"Removed reviewer Doug Hellmann.","accounts_in_message":[],"_revision_number":2},{"id":"d7c2b3fc22f2e8d761ebdeaf4b29ade2c08710c4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-02 09:14:04.000000000","message":"Patch Set 2: Code-Review+2\n\nElod\u0027s comments have been addressed, LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"f802847e03dbf6f395e270d99efa2b73cb27786d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-03-02 14:37:33.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4974f31be43c4b338168cb70cda6a5c2003344c6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-02 15:23:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe0863e9b4490a04d45b37fa4b0dabf833e5636b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-05 14:16:48.000000000","message":"Patch Set 2:\n\nbump can someone merge this please 😊","accounts_in_message":[],"_revision_number":2},{"id":"513a6c0b36257b893d790015fd97b3fcf496a3b9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-05 16:31:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nSorry again, there are some new commits for two repos, see my comments inline. :/\n\nI\u0027ll try to review quickly as soon as you\u0027ve updated this release patch. o:)\nNevertheless I suggest to not to accept changes anymore on rocky for these repositories. Also useful to remember, that after the branch gets EOL\u0027d, but the branch is not yet deleted, be careful to not to merge in new patches.","accounts_in_message":[],"_revision_number":2},{"id":"5205df09e9532beada3e1e85b94b69992e391215","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-05 16:35:47.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nthanks for checking again Elod","accounts_in_message":[],"_revision_number":2},{"id":"8066b466a8af8934a7a6cc2f3ecbcf4aeddd5efb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-05 16:45:16.000000000","message":"Patch Set 2: Code-Review-1\n\n18:40 \u003c elod\u003e marios: for the sake of completeness, please update, then I\u0027ll +2 immediately :X\n\n\nack updating now thank you","accounts_in_message":[],"_revision_number":2},{"id":"1e5fba97633166f43da3b73a6de7d575fb1e2245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-05 16:46:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d5b759e7a3a8307b34d6265faee3329c2bf12d4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:47:46.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"6ba301755b47b967a8cc8438cda0f95d928f606a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-05 16:50:21.000000000","message":"Patch Set 3: Code-Review+2\n\nThank you Marios and sorry for holding this up for so long time :S","accounts_in_message":[],"_revision_number":3},{"id":"c19a1503638629f5aa20cf72150517bc10e246ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:00:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9636b846a774497aa26c499b34cba907 : SUCCESS in 7m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b08faf2168df440ba952f5c3b5c94f28 : SUCCESS in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3babfa4b693f45089b06fe34bcc40de5 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8b19919eef215627b0eb65c5b3520ed16d32f674","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-05 17:14:47.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+2\n\u003e \n\u003e Thank you Marios and sorry for holding this up for so long time :S\n\nthank you Elod no need to apologise for doing a good job ;)","accounts_in_message":[],"_revision_number":3},{"id":"7614ad29d4fd5a9db7a8a9c646d694aa3beab371","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 17:30:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"7633747b1c4563eadadf349f4b77d22050b86abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:31:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"359d0bd3249ac4b3db82a2ee2cafe8a52fdce653","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-03-05 17:32:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"472bbfcd16bfec9eac12c62f9befa06883acebbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:40:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb9ed5ecb0594953893682d8440e0248 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b1621c43b54867821e7bd85a5c20df : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7a7abab97e94495993980f87a675ece : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":3},{"id":"db81eefde815409c37431f903f153073730ae9eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:40:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1d793e3dd13b581c47c0282c0153d27fc4621241","revisions":{"19cbafaa84ec9c4ada966494ad96ea020eca4216":{"kind":"REWORK","_number":1,"created":"2021-02-05 14:32:02.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/774244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/774244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/774244/1"}}},"commit":{"parents":[{"commit":"cb2fe6890871295b59fce4b44572cf6de32c09e4","subject":"Merge \"Release panko for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb2fe6890871295b59fce4b44572cf6de32c09e4"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-05 14:25:19.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-05 14:25:19.000000000","tz":120},"subject":"Tag stable/rocky tripleo repos as EOL","message":"Tag stable/rocky tripleo repos as EOL\n\nAs discussed at [1],[2] this tags the last commits to the\nstable/rocky branch as rocky-eol.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019860.html\n\nChange-Id: I3523b079b457b8f94780bd7c8f6e616f81f6eb40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19cbafaa84ec9c4ada966494ad96ea020eca4216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19cbafaa84ec9c4ada966494ad96ea020eca4216"}]},"branch":"refs/heads/master"},"bfb91b1323f79e809aa620b3f4dbb5aed8ca9c36":{"kind":"REWORK","_number":2,"created":"2021-03-01 09:41:55.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/774244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/774244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/774244/2"}}},"commit":{"parents":[{"commit":"1fea88eccf811e915f2116b2fc00313dd0282aaf","subject":"Merge \"ldappool 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fea88eccf811e915f2116b2fc00313dd0282aaf"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-05 14:25:19.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-03-01 09:41:15.000000000","tz":120},"subject":"Tag stable/rocky tripleo repos as EOL","message":"Tag stable/rocky tripleo repos as EOL\n\nAs discussed at [1],[2] this tags the last commits to the\nstable/rocky branch as rocky-eol.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019860.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html\n\nChange-Id: I3523b079b457b8f94780bd7c8f6e616f81f6eb40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfb91b1323f79e809aa620b3f4dbb5aed8ca9c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfb91b1323f79e809aa620b3f4dbb5aed8ca9c36"}]},"branch":"refs/heads/master"},"1d793e3dd13b581c47c0282c0153d27fc4621241":{"kind":"REWORK","_number":3,"created":"2021-03-05 16:46:32.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/774244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/774244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/774244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/774244/3"}}},"commit":{"parents":[{"commit":"c9e96c9c20280184090c82b46571ee6c69f0cadc","subject":"Merge \"Adding a tool to catch projects that missed branching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e96c9c20280184090c82b46571ee6c69f0cadc"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-05 14:25:19.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-03-05 16:43:59.000000000","tz":120},"subject":"Tag stable/rocky tripleo repos as EOL","message":"Tag stable/rocky tripleo repos as EOL\n\nAs discussed at [1],[2] this tags the last commits to the\nstable/rocky branch as rocky-eol.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019338.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019860.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020274.html\n\nChange-Id: I3523b079b457b8f94780bd7c8f6e616f81f6eb40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d793e3dd13b581c47c0282c0153d27fc4621241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d793e3dd13b581c47c0282c0153d27fc4621241"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
