)]}'
{"id":"openstack%2Freleases~774683","triplet_id":"openstack%2Freleases~master~I5bfd701e6059d2dfdc645bb5f2c7a4da249ad1f2","project":"openstack/releases","branch":"master","topic":"switch-to-rc","hashtags":[],"change_id":"I5bfd701e6059d2dfdc645bb5f2c7a4da249ad1f2","subject":"Propose a release model change for Vitrage-dashboard","status":"ABANDONED","created":"2021-02-09 15:34:41.000000000","updated":"2021-02-18 16:34:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"598f101bfc77d6e812329bf06cb30666de815c29","_number":774683,"virtual_id_number":774683,"owner":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-09 15:56:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-02-10 09:41:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 15:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-09 15:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 16:20:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"bb90e65486e039f5a0acb3be54b93414e6d7bda5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2021-02-09 15:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfda37167cc34a0afc8533f77f4e7bad679e322f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 15:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf7c70060cd4cf8b494c20daeb2b0fc : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e53edea26e2843178ea3829f41034d8f : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a86926ec4c514bc59dc209460e3c8fa0 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8de9d4eb307d55cbaeca57684323c466b3b4bf92","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-02-10 09:41:58.000000000","message":"Patch Set 1: Code-Review-1\n\nreleased new versions of vitrage and vitrage-dashboard\n\nhttps://review.opendev.org/c/openstack/releases/+/774783","accounts_in_message":[],"_revision_number":1},{"id":"598f101bfc77d6e812329bf06cb30666de815c29","tag":"autogenerated:gerrit:abandon","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-18 16:34:30.000000000","message":"Abandoned\n\nVitrage Dashboard will not be changing model in Wallaby","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3ab4a0ba1d487202ca6735c85f0871f25a7bd93","revisions":{"d3ab4a0ba1d487202ca6735c85f0871f25a7bd93":{"kind":"REWORK","_number":1,"created":"2021-02-09 15:34:41.000000000","uploader":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"ref":"refs/changes/83/774683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/774683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/774683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/774683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/774683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/774683/1"}}},"commit":{"parents":[{"commit":"143b935aee0547610cb4d7e320b6ead6a2b33181","subject":"Merge \"New bugfix release for train oslo.serialization.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/143b935aee0547610cb4d7e320b6ead6a2b33181"}]}],"author":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-02-09 15:33:05.000000000","tz":-300},"committer":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-02-09 15:33:05.000000000","tz":-300},"subject":"Propose a release model change for Vitrage-dashboard","message":"Propose a release model change for Vitrage-dashboard\n\nDeliverables was only released once during the last cycle, and have not done a release yet\n\nIt could be the right time to switch to the cycle-with-rc model.\nPTLs and release liaisons please:\n- immediately release an intermediary release (and -1 the proposed change)\n- confirm the release model change (+1 the proposed change)\n- stay uncertain for this cycle of how many releases will be made,\n  but acknowledge that they need to do a release before RC1\n  (-1 the proposed change)\n\nChange-Id: I5bfd701e6059d2dfdc645bb5f2c7a4da249ad1f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3ab4a0ba1d487202ca6735c85f0871f25a7bd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3ab4a0ba1d487202ca6735c85f0871f25a7bd93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
