)]}'
{"id":"openstack%2Freleases~775375","triplet_id":"openstack%2Freleases~master~Ifd402f5ab863754e8a3f645fcfb4915f72cf13a6","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ifd402f5ab863754e8a3f645fcfb4915f72cf13a6","subject":"Add TLS support to oslo.cache","status":"MERGED","created":"2021-02-12 12:35:09.000000000","updated":"2021-02-16 09:47:55.000000000","submitted":"2021-02-16 09:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"775375","meta_rev_id":"3fb8239ab36a71861e1c1b022bd3ddc2cbb7f96b","_number":775375,"virtual_id_number":775375,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 09:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-02-16 09:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-02-12 21:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-02-12 13:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-12 12:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-02-16 09:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-02-12 12:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 12:36:39.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-02-12 12:36:39.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-12 12:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 12:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 21:19:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-02-16 09:32:12.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"0667adbec7bece4ac2872b763a47cdabcbd81347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-12 12:35:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a369a6b8b5ed9fcded255cd9596d6a93fdce696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-12 12:35:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"64657000ac62d75ef0643b9d28ee35c4397e0294","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-12 12:39:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"44292e96921398fb486a122cf728a02efb02899f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 12:41:13.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"073e649cb0e9d190425999c80ea67df2a1470f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 12:42:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6741e44d168c47ed97e95f57de7d1d13 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e98d99c9ae154f14ac0e6b13228c248b : SUCCESS in 4m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a9c6dfc3de6d4563a0d196fe0033fa4c : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"816b8bfc590b92b01bd8ee49d0e52bc7b230ba80","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-12 13:50:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"88d2791d1857feb50981e9cfc7891ada79a61639","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-12 21:19:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLooks OK to me:\n* version bumps are OK (we cannot bump minor version as those are already used on newer branches)\n* validator job passed (warnings are due to the mentioned issue)\n* list-changes-results.log details looks OK to me\n\nLeaving the release to Hervé.","accounts_in_message":[],"_revision_number":2},{"id":"2f417508d60f0cc757263378c2d9fa6c783fdf2a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-16 09:32:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbd2ea018be1e40f6fb741558b10637d2297aab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 09:33:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5eaee1fc734e4df3c40c5bcbfa09f335b0d3db56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 09:47:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694a1b192530492991b125ef4bfe15b0 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b2b903db594e73b45898ff36fe7d1a : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2276c4e424b4042b6d3d1d6b3d7881b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3fb8239ab36a71861e1c1b022bd3ddc2cbb7f96b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 09:47:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"412fe4a39f2da215e8dbccb98f092f3bd2ee7452","revisions":{"9296030ba83adb6ff2184e6991fd5495c4920bd6":{"kind":"REWORK","_number":1,"created":"2021-02-12 12:35:09.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/75/775375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/775375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/775375/1"}}},"commit":{"parents":[{"commit":"c9e3f48528b264befd835782bef91b56c020b245","subject":"Merge \"Add Extra ATC during R-7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e3f48528b264befd835782bef91b56c020b245"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-12 12:33:16.000000000","tz":60},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-12 12:33:16.000000000","tz":60},"subject":"Add TLS support to oslo.cache","message":"Add TLS support to oslo.cache\n\nContains new releases of oslo.cache for both ussuri and train branches.\n\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\nChange-Id: Ifd402f5ab863754e8a3f645fcfb4915f72cf13a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9296030ba83adb6ff2184e6991fd5495c4920bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9296030ba83adb6ff2184e6991fd5495c4920bd6"}]},"branch":"refs/heads/master"},"412fe4a39f2da215e8dbccb98f092f3bd2ee7452":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-12 12:35:25.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/75/775375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/775375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/775375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/775375/2"}}},"commit":{"parents":[{"commit":"c53da6d7bbd7abbfc50c7f67da1adec763fed5c0","subject":"New bugfix and feature releases for oslo.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c53da6d7bbd7abbfc50c7f67da1adec763fed5c0"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-12 12:33:16.000000000","tz":60},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2021-02-12 12:35:25.000000000","tz":0},"subject":"Add TLS support to oslo.cache","message":"Add TLS support to oslo.cache\n\nContains new releases of oslo.cache for both ussuri and train branches.\n\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\nChange-Id: Ifd402f5ab863754e8a3f645fcfb4915f72cf13a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/412fe4a39f2da215e8dbccb98f092f3bd2ee7452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/412fe4a39f2da215e8dbccb98f092f3bd2ee7452"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
