)]}'
{"id":"openstack%2Freleases~776496","triplet_id":"openstack%2Freleases~master~I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7","project":"openstack/releases","branch":"master","topic":"ensure-branching","hashtags":[],"change_id":"I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7","subject":"Adding a tool to catch projects that missed branching","status":"MERGED","created":"2021-02-18 16:59:34.000000000","updated":"2021-03-05 14:20:39.000000000","submitted":"2021-03-05 14:20:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"776496-ensure-branching","meta_rev_id":"f943a0eb2d50d2959cd480ddee4db6f83fafddbb","_number":776496,"virtual_id_number":776496,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 14:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-04 10:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-05 14:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-03-05 14:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 17:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 10:16:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-03-05 14:09:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"566914538420a3a70b30148dafb1f3024c98ea41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 16:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b2bb3154deb18158ac5a85e6ab17423acb451bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:15:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f72e2cfd3d44c4afbf9266fcee9b6c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00077265eac046d9ac490a44ffd0a038 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f772dd875a894f73a1f690ce8b2930e7 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2df2d8a9ecf4a089834d3ade80c31b7 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/336ba83913ea4a6191ec52c4ce84be85 : SUCCESS in 4m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/725ff12281b749d2be2bad3de6221c5c : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a59b69a3037411e8965552b5c541c8b : SUCCESS in 7m 01s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f320126abdc04a00bc1eda8673661587 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"04430562d33571b6ef15237eb8b108cf88008301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 18:11:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e365b58af5591b6213836df525b2634d124f5d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:23:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152b62689e014c419db320153c19d8a6 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc773f80e0d44bf9edcca25987a295a : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a86e34f4b8b14f82abda9b4c4ec44351 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644656b5a13144fd81dded14c288536d : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e73ed37b8374496b0c1dcf217e9d9fb : SUCCESS in 5m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/11b9203279cf477db00fb255830c8172 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9eb1cdbf8ff411d9ae97702333f1a17 : SUCCESS in 4m 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f37673e183ea4a78a96373ada3e4d85c : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3cfce0e4b8b6ba9472d9353a8549edbb6dd05b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 18:51:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87897d3579d3c8b24f70e476b197da91e4b89ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 19:18:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c420f492ff43539cf3a43d78b76f0d : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/076a501468894c988515498c65008f9d : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55c79953dd9c453283d840f471bb654b : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac3494c4b2c4fae9f82560cbb925621 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5093ecb5fc46496ea7a6a302f6e8148d : SUCCESS in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bad120049fa84fba9dacf0298d03f357 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8691c50f0b741d8a7f4eb39ea0892c8 : SUCCESS in 4m 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1928c6bb53224531b1d2d3bd604a8323 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"6b6e4683e0ae25c1b9da55720acd83762b68afda","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-22 10:16:56.000000000","message":"Patch Set 3: Code-Review-1\n\nLooks great, but you need to update doc/source/reference/using.rst too.","accounts_in_message":[],"_revision_number":3},{"id":"6d2e207da0a1af914015bd86325629b35cafb256","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-25 15:19:38.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e Looks great, but you need to update doc/source/reference/using.rst too.\n\nYes, I had previously planned to update it and update our process to add usages within our schedule around each trailing deadline to make sure everything is correct.\n\nI will suggest an update next week.","accounts_in_message":[],"_revision_number":3},{"id":"74ce5e872fa08f3c327fdb2cb509c3a3af7895cc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-25 15:20:51.000000000","message":"Patch Set 3: Workflow-1\n\nStill WIP","accounts_in_message":[],"_revision_number":3},{"id":"71bffe53112cdd18827806ffb355e8d892ce4a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 14:59:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0308d544b3061041121d579782d302b09219c890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:17:13.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb93957951b4f028b90494665b94cc4 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a64a94686cc5475bb4eb3aced98be8f4 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cf22b4d92e482b8ddb15759173fa84 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba73677b757e43d6ab956dc23856b122 : FAILURE in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fbde3c9f2f24db8b135edd273f72075 : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5189875a805d4eefb53c97a051b3d249 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f1e8b8a60b8444090f071dc515e1886 : SUCCESS in 6m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/00607d6b5cb242488eef621accd1fa1f : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":4},{"id":"64a9f2edc63cf56084ea4e22ca8ede32d8d8d684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:17:53.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb93957951b4f028b90494665b94cc4 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a64a94686cc5475bb4eb3aced98be8f4 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cf22b4d92e482b8ddb15759173fa84 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba73677b757e43d6ab956dc23856b122 : FAILURE in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fbde3c9f2f24db8b135edd273f72075 : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5189875a805d4eefb53c97a051b3d249 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f1e8b8a60b8444090f071dc515e1886 : SUCCESS in 6m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/00607d6b5cb242488eef621accd1fa1f : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":4},{"id":"5833fac36e5c21763169fbbc715ab14de9ba5368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:18:43.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb93957951b4f028b90494665b94cc4 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a64a94686cc5475bb4eb3aced98be8f4 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cf22b4d92e482b8ddb15759173fa84 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba73677b757e43d6ab956dc23856b122 : FAILURE in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fbde3c9f2f24db8b135edd273f72075 : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5189875a805d4eefb53c97a051b3d249 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f1e8b8a60b8444090f071dc515e1886 : SUCCESS in 6m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/00607d6b5cb242488eef621accd1fa1f : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":4},{"id":"fe558bfbc77779a08b331f4962bc15c1b0d61239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 12:51:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"840978ddf240790a7d3d6f7db08271c11fb6e0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:10:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d913913eb5b5497799c83c92b6c6f52c : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d9ee89adb21478c88d4cefa839be984 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef800c7cc6f74ef58e664fe718c68069 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fb07daf9ed4632876c1f7378695092 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d540780096243a79a6594f2b3e854b1 : SUCCESS in 7m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5517d18f786f4b8ea81dee6bbde96191 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9588874dbd04432c8d8f528352c092bc : SUCCESS in 6m 04s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/01b346b0b8814752b42022ff96a9bf3e : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":5},{"id":"113e8a7971ac906eb0f86674f6254f4a165a2441","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-04 10:32:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"367f0ea2450ed58a3e40b0f3e94f4877632ca1d2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-05 14:09:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)\n\nThere are some nits, otherwise the script works well and will be useful in the coming days.","accounts_in_message":[],"_revision_number":5},{"id":"72c5bd18fd7fca14706dfd898483995cbb90aba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:09:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad5e834d4111788631c8cd494d7b99dfc87e73d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:20:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a3f40543b54c1a8be3f93b3624d4d2 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b0c039924254dfaaaff41df76cd2f70 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/294928dfed304655b360bfffbb098e0a : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2321b546c6414153b4fac2394ff46379 : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/506465f7d3704117b11c016d86146919 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/611e05a97e464b0e99f7566b5b188ef1 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":5},{"id":"f943a0eb2d50d2959cd480ddee4db6f83fafddbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c42fa888659b0ad7880c6c9bfac4734bfa56b825","revisions":{"c95b6b6348cdf3bb53ecc57905daa73a0fda9d0c":{"kind":"REWORK","_number":1,"created":"2021-02-18 16:59:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/776496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/776496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/776496/1"}}},"commit":{"parents":[{"commit":"924baff62ebb9847a47298a69580a87f852ea562","subject":"Merge \"Move tripleo-ipsec repo to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/924baff62ebb9847a47298a69580a87f852ea562"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:56:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:59:01.000000000","tz":60},"subject":"[WIP] Adding a tool to catch projects that missed branching","message":"[WIP] Adding a tool to catch projects that missed branching\n\nThis tool will be used in doc addition around the branching period\nto ensure that we don\u0027t miss to create stable branches for project.\n\nChange-Id: I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c95b6b6348cdf3bb53ecc57905daa73a0fda9d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c95b6b6348cdf3bb53ecc57905daa73a0fda9d0c"}]},"branch":"refs/heads/master"},"8a9e5500de5d4350a26baca960e7abed1a54db81":{"kind":"REWORK","_number":2,"created":"2021-02-18 18:11:03.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/776496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/776496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/776496/2"}}},"commit":{"parents":[{"commit":"924baff62ebb9847a47298a69580a87f852ea562","subject":"Merge \"Move tripleo-ipsec repo to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/924baff62ebb9847a47298a69580a87f852ea562"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:56:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 18:10:44.000000000","tz":60},"subject":"[WIP] Adding a tool to catch projects that missed branching","message":"[WIP] Adding a tool to catch projects that missed branching\n\nThis tool will be used in doc addition around the branching period\nto ensure that we don\u0027t miss to create stable branches for project.\n\nChange-Id: I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a9e5500de5d4350a26baca960e7abed1a54db81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a9e5500de5d4350a26baca960e7abed1a54db81"}]},"branch":"refs/heads/master"},"76a700a78d7c660b036bba89ec7e8123ce14e2e7":{"kind":"REWORK","_number":3,"created":"2021-02-18 18:51:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/776496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/776496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/776496/3"}}},"commit":{"parents":[{"commit":"924baff62ebb9847a47298a69580a87f852ea562","subject":"Merge \"Move tripleo-ipsec repo to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/924baff62ebb9847a47298a69580a87f852ea562"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:56:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 18:49:59.000000000","tz":60},"subject":"Adding a tool to catch projects that missed branching","message":"Adding a tool to catch projects that missed branching\n\nThis tool will be used in doc addition around the branching period\nto ensure that we don\u0027t miss to create stable branches for project.\n\nAlso this tooling should be used around each trailing deadline to ensure\nto not forget a trailing projects. Trailing projects are those who can\neasily meet the conditions that lead to forget to branch them.\n\nChange-Id: I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76a700a78d7c660b036bba89ec7e8123ce14e2e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76a700a78d7c660b036bba89ec7e8123ce14e2e7"}]},"branch":"refs/heads/master"},"4e3766f058a129ac9852b65300c28d7eb3372dfc":{"kind":"REWORK","_number":4,"created":"2021-03-01 14:59:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/776496/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/776496/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/776496/4"}}},"commit":{"parents":[{"commit":"924baff62ebb9847a47298a69580a87f852ea562","subject":"Merge \"Move tripleo-ipsec repo to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/924baff62ebb9847a47298a69580a87f852ea562"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:56:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-01 14:58:46.000000000","tz":60},"subject":"Adding a tool to catch projects that missed branching","message":"Adding a tool to catch projects that missed branching\n\nThis tool will be used in doc addition around the branching period\nto ensure that we don\u0027t miss to create stable branches for project.\n\nAlso this tooling should be used around each trailing deadline to ensure\nto not forget a trailing projects. Trailing projects are those who can\neasily meet the conditions that lead to forget to branch them.\n\nAdding usage of this tooling in our process to ensure to check that\npoint once a time at each new series.\n\nChange-Id: I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e3766f058a129ac9852b65300c28d7eb3372dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e3766f058a129ac9852b65300c28d7eb3372dfc"}]},"branch":"refs/heads/master"},"c42fa888659b0ad7880c6c9bfac4734bfa56b825":{"kind":"REWORK","_number":5,"created":"2021-03-03 12:51:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/776496/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/776496/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/776496/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/776496/5"}}},"commit":{"parents":[{"commit":"924baff62ebb9847a47298a69580a87f852ea562","subject":"Merge \"Move tripleo-ipsec repo to independent release model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/924baff62ebb9847a47298a69580a87f852ea562"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 16:56:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-03 12:51:12.000000000","tz":60},"subject":"Adding a tool to catch projects that missed branching","message":"Adding a tool to catch projects that missed branching\n\nThis tool will be used in doc addition around the branching period\nto ensure that we don\u0027t miss to create stable branches for project.\n\nAlso this tooling should be used around each trailing deadline to ensure\nto not forget a trailing projects. Trailing projects are those who can\neasily meet the conditions that lead to forget to branch them.\n\nAdding usage of this tooling in our process to ensure to check that\npoint once a time at each new series.\n\nChange-Id: I2a8bd25ecfe5bb1bde5af16b08f897a5bdc11cb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c42fa888659b0ad7880c6c9bfac4734bfa56b825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c42fa888659b0ad7880c6c9bfac4734bfa56b825"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
