)]}'
{"id":"openstack%2Freleases~776636","triplet_id":"openstack%2Freleases~master~Ifeee0d6b5422c8f36843c1c83196f8a7ec3e7d99","project":"openstack/releases","branch":"master","topic":"missing-stable-branch","hashtags":[],"change_id":"Ifeee0d6b5422c8f36843c1c83196f8a7ec3e7d99","subject":"Tag and create missing ussuri branch os-refresh-config","status":"MERGED","created":"2021-02-19 12:44:23.000000000","updated":"2021-03-11 18:03:31.000000000","submitted":"2021-03-11 18:03:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"776636-missing-stable-branch","meta_rev_id":"1b0f69decf3586a5be2019c42e5f36a1c070a1a8","_number":776636,"virtual_id_number":776636,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 18:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-03 15:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-03-11 10:02:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-11 17:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-11 17:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-03-11 17:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-03 15:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 12:45:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 12:51:32.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-02-19 12:51:32.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-19 13:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-03-01 09:41:37.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-03-01 14:34:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-03-11 10:02:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-03-11 17:50:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"84d25784a929c76b864e11e21053846f38c8d636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-19 12:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25d719510fadf5356d0d8c137526971cdea3f0df","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:45:34.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0e5413de379c5909c57dc2d3c9bccec17a36b76d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-19 12:51:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e0cc4a1e1297d841700a805396f43c2e169ccff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c9a0a2d5bd42a38a0cad94b88ff517 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/888f8946170246efb844fe3f05e48fa1 : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f75a791f2fec40c7821d1695f47fd24f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a30257c60e318291e94d698c22d2d0b90d7e2684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-19 16:14:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3d8db0a6cc797d2d2c71bb7cbfb9c63a65ae4b60","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:15:31.000000000","message":"Patch Set 2: -PTL-Approved","accounts_in_message":[],"_revision_number":2},{"id":"7257bd0e2e159e849363eb84f8d6177387296a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:28:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066a54927cf845889a06d5fa1f28dd69 : POST_FAILURE in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d5df2d0e21648c488e6199d7dacbaf6 : POST_FAILURE in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f94246e3c5f246a6802ad8f3f462f30e : POST_FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"44ec7994fb62bec6c829271bd4863883c6faba83","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-22 14:39:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"502a522e5a9e8f259301c9e0135c3359f04cc22c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-01 09:41:37.000000000","message":"Patch Set 2:\n\nrecheck no logs posted","accounts_in_message":[],"_revision_number":2},{"id":"e512b043ec70cd77d9ad9469f9c183eeeac12e01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:16:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e95adbe6e964b4088bd68feea30957a : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/774ff22ccf96418298cc575a4f6e0251 : SUCCESS in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/04096b3c559340bcb52817436771021b : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cfd181398e57dd41d8799e9e8655505768e6f5d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-01 14:34:54.000000000","message":"Patch Set 2:\n\nOn the one hand, it seems awfully late to be adding something to the ussuri release. With the switch to this being an independent deliverable, is a cycle-based stable branch still needed here?\n\nhttps://opendev.org/openstack/releases/commit/78d0577401838b7066341572346f7e7e99640d40","accounts_in_message":[],"_revision_number":2},{"id":"4ac066529e946d0e8367eccbc1f7c529ef894c62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 15:11:04.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2:\n\u003e \n\u003e On the one hand, it seems awfully late to be adding something to the ussuri release. With the switch to this being an independent deliverable, is a cycle-based stable branch still needed here?\n\u003e \n\u003e https://opendev.org/openstack/releases/commit/78d0577401838b7066341572346f7e7e99640d40\n\nIndeed this project is now independent so technically I think it could be ignored, however I think it\u0027s worth to catch this case now by creating the branch at the right point to avoid this in future (hypothetical). We have suffered enough from similar issues over the past few weeks to secure it now while it is still hot rather than ignore it and to face the same situation and reask the same questions in few months.\n\nThoughts?","accounts_in_message":[],"_revision_number":2},{"id":"792c7f62c9d3d8f82d7c8ff4023cfe245053ead0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 12:40:05.000000000","message":"Patch Set 2: Code-Review+1\n\nHere is my vote to give my position about this topic.\n\nI\u0027ll be happy to discuss with you if we want to continue or not and to measure the pro and con.\n\nFor me the rationale behind these changes are enough worthful to allow us to continue.\n\nNotice that I only +1 to avoid to hide possible -1.","accounts_in_message":[],"_revision_number":2},{"id":"fb9b8d96697158cd625696536359d2895b800ab0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 13:16:20.000000000","message":"Patch Set 2:\n\nHere is a more or less related topic https://bugs.launchpad.net/tripleo-ipsec/+bug/1917327\n\nIn the current case (this patch) we didn\u0027t released the same changes twice but we met more or less the same condition here, so I think we should consider that point.\n\nAs Murphy said, anything that can go wrong will go wrong.\n\nThe paint of this topic is still enough fresh so I think we should address this point now rather than abandonned it for now and reask the same question in few months. For now we know what\u0027s go wrong.\n\nNotice that the pbr issue is now solved by https://review.opendev.org/c/openstack/releases/+/776206/\n\nIf we decide to continue here we will avoid the pbr problem.","accounts_in_message":[],"_revision_number":2},{"id":"0122b87c7a2084e19333cab6a79d972a1032f163","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-03 15:25:14.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e On the one hand, it seems awfully late to be adding something to the ussuri release. With the switch to this being an independent deliverable, is a cycle-based stable branch still needed here?\n\u003e \n\u003e https://opendev.org/openstack/releases/commit/78d0577401838b7066341572346f7e7e99640d40\n\nHi Sean\n\nyeah indeed - I mean, to be clear, this isn\u0027t a case of us actually *needing* the ussuri branch. We aren\u0027t using the project since before the U cycle (hence why we didn\u0027t notice about this missing branch). \n\nI posted this patch on a prompt from Herve - it was my understanding that this would help to fix something in the release jobs that was broken by the missing branch.\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"8f45e65cfc446a2a6535b3c4950685b71106b886","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-03 15:26:55.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Here is a more or less related topic https://bugs.launchpad.net/tripleo-ipsec/+bug/1917327\n\u003e \n\u003e In the current case (this patch) we didn\u0027t released the same changes twice but we met more or less the same condition here, so I think we should consider that point.\n\u003e \n\u003e As Murphy said, anything that can go wrong will go wrong.\n\u003e \n\u003e The paint of this topic is still enough fresh so I think we should address this point now rather than abandonned it for now and reask the same question in few months. For now we know what\u0027s go wrong.\n\u003e \n\u003e Notice that the pbr issue is now solved by https://review.opendev.org/c/openstack/releases/+/776206/\n\u003e \n\u003e If we decide to continue here we will avoid the pbr problem.\n\nthanks Herve for your work here.\n\n\nAs I just replied to Sean - we don\u0027t actually *need* a stable/ussuri for tripleo, but, I am happy to land this change if it will help us to avoid any release related issues in the future.\n\nthank you","accounts_in_message":[],"_revision_number":2},{"id":"b40d8dfa7af828dd97ddcb5eb5bdcbc52dd46301","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-03 15:27:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc66abaf63ebbacafbce2445e5dcac199dba0820","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:33:41.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"cb1a76679c72dd639cbf19a024e95ae3200c0ee1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 17:33:09.000000000","message":"Patch Set 2:\n\nAlso I think that we should consider EM/EOL things during our discussion about this kind of problem.\n\nThey risk to emerge again at this time","accounts_in_message":[],"_revision_number":2},{"id":"b8ee6ede526d756b3807a014c8d372eaa95ece6b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-11 10:02:21.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027m fine with it","accounts_in_message":[],"_revision_number":2},{"id":"0a6e493fbd59a63321d6fcd80e8d8b5dc3509f2c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 13:06:46.000000000","message":"Patch Set 2:\n\nWe didn\u0027t get new feedback since 10 days, I think we can go ahead.\nAdded to our meeting agenda to move further and give a last chance to people to raise hand.","accounts_in_message":[],"_revision_number":2},{"id":"3befa625045f2ae7366dbc70d17e4aa4f1cfce57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 17:48:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be90553b294450f3c26ca8d1c71e48ef920b7dd9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-11 17:50:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nApproving as discussed on today\u0027s release meeting.","accounts_in_message":[],"_revision_number":2},{"id":"9c8313df95c6b73ed57598d6258daf5c3adfb376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:50:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc2bb1a8c903a8ee1f509476bf64a4152112cb03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:03:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3253adb2924df18a6db72ecd77de2b : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b51d4bbc6e42379c19f93a29c93d99 : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc155844feb54dba9ff234ad6f5660cd : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"1b0f69decf3586a5be2019c42e5f36a1c070a1a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:03:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6495f63d8dc533f6e050381ff6bee58f11dd90b","revisions":{"3a7c9658561ce94f2a160d93af525c83f27cf43c":{"kind":"REWORK","_number":1,"created":"2021-02-19 12:44:23.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/36/776636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/776636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/776636/1"}}},"commit":{"parents":[{"commit":"bfb96284ff9b2236e0c68a8701ecc032da3aeaf2","subject":"automaton 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfb96284ff9b2236e0c68a8701ecc032da3aeaf2"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-19 12:37:56.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-19 12:43:18.000000000","tz":120},"subject":"Tag and create missing ussuri branch os-refresh-config","message":"Tag and create missing ussuri branch os-refresh-config\n\nAs discussed at [1] this tags 11.0.0 and uses it to create missing\nussuri branch for os-refresh-config.\n\n[1] http://eavesdrop.openstack.org/irclogs/%23tripleo/%23tripleo.2021-02-18.log.html#t2021-02-18T16:03:22\n\nChange-Id: Ifeee0d6b5422c8f36843c1c83196f8a7ec3e7d99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a7c9658561ce94f2a160d93af525c83f27cf43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a7c9658561ce94f2a160d93af525c83f27cf43c"}]},"branch":"refs/heads/master"},"a6495f63d8dc533f6e050381ff6bee58f11dd90b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-19 16:14:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/36/776636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/776636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/776636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/776636/2"}}},"commit":{"parents":[{"commit":"80652b523214ab7215928368246fbf726b78a645","subject":"Ensure that we can build sdist from requested tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-02-19 12:37:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-19 16:12:27.000000000","tz":60},"subject":"Tag and create missing ussuri branch os-refresh-config","message":"Tag and create missing ussuri branch os-refresh-config\n\nAs discussed at [1] this tags 11.0.0 and uses it to create missing\nussuri branch for os-refresh-config.\n\n[1] http://eavesdrop.openstack.org/irclogs/%23tripleo/%23tripleo.2021-02-18.log.html#t2021-02-18T16:03:22\n\nChange-Id: Ifeee0d6b5422c8f36843c1c83196f8a7ec3e7d99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6495f63d8dc533f6e050381ff6bee58f11dd90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6495f63d8dc533f6e050381ff6bee58f11dd90b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
