)]}'
{"id":"openstack%2Freleases~778965","triplet_id":"openstack%2Freleases~master~I55b40d00a58128a652560aba86d0be5712712ad9","project":"openstack/releases","branch":"master","topic":"cwi-client-final","hashtags":[],"change_id":"I55b40d00a58128a652560aba86d0be5712712ad9","subject":"Release final python-manilaclient for Wallaby","status":"MERGED","created":"2021-03-05 15:20:39.000000000","updated":"2021-03-15 20:29:47.000000000","submitted":"2021-03-15 20:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778965-cwi-client-final","meta_rev_id":"10c950247f8aa6499c4161b840373ec211ee599e","_number":778965,"virtual_id_number":778965,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 20:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-03-15 20:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-03-15 20:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-15 20:12:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 15:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 15:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 16:21:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-05 16:22:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"9ee1d6e64c32cd048b91af4e95f884a6e8d602e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 15:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"439e67b7b6ff8ab7f876cef180ad6027f36b8cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:40:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823b5170fd3046a4aebf46fd7379ee71 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/774fcf025041456dbf9b8c9d6e27112e : SUCCESS in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/172b03a10a72471ba1d89cc98a197677 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9e9cd86a30d5d971529c7dac4b4040ea560ab525","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-05 18:27:15.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease hold off, we have a few more changes pending for this project which we hope to merge by 11th March","accounts_in_message":[],"_revision_number":1},{"id":"9f45ed6f5fa1224e61f19d18008bb45b23370f6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-11 13:11:57.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"f5eb424856f93e0c39a02de0be42fbf4a4114053","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-11 23:17:46.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Any update?\n\nUnfortunately not :( \nWe\u0027re tracking one last change to merge this cycle:\n\nhttps://review.opendev.org/c/openstack/python-manilaclient/+/775020\n\nI\u0027d like some more time to get this in - we\u0027ll try to wrap this up tomorrow/Mar 12. Would that be possible?","accounts_in_message":[],"_revision_number":1},{"id":"5d890f9ac65bdd6e65e006e5907ad25c65235eeb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 09:02:44.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Any update?\n\u003e \n\u003e Unfortunately not :( \n\u003e We\u0027re tracking one last change to merge this cycle:\n\u003e \n\u003e https://review.opendev.org/c/openstack/python-manilaclient/+/775020\n\u003e \n\u003e I\u0027d like some more time to get this in - we\u0027ll try to wrap this up tomorrow/Mar 12. Would that be possible?\n\nYes I think we can wait for early next week (monday).","accounts_in_message":[],"_revision_number":1},{"id":"9c5fb5f1b958dbd5c9d6aad7709980960a51eb39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 08:23:28.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"218a640d773177725ed1bed9cb591aa27d6a44d0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 15:02:15.000000000","message":"Patch Set 1:\n\nI `recheck` twice your patch (https://review.opendev.org/c/openstack/python-manilaclient/+/775020) I hope it will pass. \n\nIf this patch continue to fail we\u0027ll need to force this release ASAP to allow the requirements team to freeze.","accounts_in_message":[],"_revision_number":1},{"id":"4b8b7554fff16263deca86890200f610e4788219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-15 20:11:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a082b6c9a9ba3ffee1772dacc136e2513723478","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:12:28.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"2c71576783b67d0944e5bd9706cce6b2ef32913b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:14:20.000000000","message":"Patch Set 2: Code-Review+2\n\nRebased on master\u0027s HEAD LGTM waiting for gates","accounts_in_message":[],"_revision_number":2},{"id":"aba4186b4a1dce974db39758c7ae355292fa53fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:18:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b5475f6d9b404785c0e3eacc8be0df : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/778de4b719e74536908cc270d39b4bf9 : SUCCESS in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5916b30851ee418ca0bdcc25c8c1c90b : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"98bb26fc4c716c6df50fb0c58d070cd4c33f1f65","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:19:12.000000000","message":"Patch Set 2: Workflow+1\n\nFast approving this patch. This the last missing freeze for lib.","accounts_in_message":[],"_revision_number":2},{"id":"5d6b3d0477b0b75c9665b8f13568340e76a4e12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:20:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b045d5854a4b77e6c480e4c552deb2d6f59c8774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:29:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe5617276154e71b5c01275da8d559c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb1fcc62e7a42bf919057fdf5bc0c3d : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a482ba4af5274f228c0d3980864a3107 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"10c950247f8aa6499c4161b840373ec211ee599e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fc1604520aad4eb43f5e547963ea16bee620a2c","revisions":{"7e8e2b55a0225da9e0f7d27867018ac1cb4a91b9":{"kind":"REWORK","_number":1,"created":"2021-03-05 15:20:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/65/778965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/778965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/778965/1"}}},"commit":{"parents":[{"commit":"c9e96c9c20280184090c82b46571ee6c69f0cadc","subject":"Merge \"Adding a tool to catch projects that missed branching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e96c9c20280184090c82b46571ee6c69f0cadc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-05 15:20:25.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-05 15:20:25.000000000","tz":60},"subject":"Release final python-manilaclient for Wallaby","message":"Release final python-manilaclient for Wallaby\n\nThis proposed a release for python-manilaclient to get any final merged changes\nout prior to the lib freeze for Wallaby going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Monday, March 11.\n\n$ git log --oneline --no-merges 2.4.0..4f5d596\nb5b8bdb Add \"--wait\" option for force-deleting a share/snapshot/share-instance\n9bc7cec Change public share tests\nc225f0a [OSC] Implement user messages commands\nca65447 tox miniversion update to 3.1.1\na34aca6 Add profiling support to manilaclient\nfc5b6c5 Forbid users to create shares with the name \"None\"\n17e5d63 [OSC] Implement Share Revert Command\nc7bcec2 Implement openstack share properties show command\nf1abf30 [OSC] Add the option to explicitly set share status\n08fb635 Tighten asserts in some wait tests\n153aa61 Add a waiter to share extend/shrink\n3c65e6f [OSC] Implement Share Export Location Commands\nd832baa [OSC] Implement the second part of share snapshots commands\nbf3e7cb [OSC] Implement Share Adopt \u0026 Abandon Commands\nb77a8c1 Replace deprecated UPPER_CONSTRAINTS_FILE variable\n07898fa Fixes create share from snapshot in OSC\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I55b40d00a58128a652560aba86d0be5712712ad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e8e2b55a0225da9e0f7d27867018ac1cb4a91b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e8e2b55a0225da9e0f7d27867018ac1cb4a91b9"}]},"branch":"refs/heads/master"},"7fc1604520aad4eb43f5e547963ea16bee620a2c":{"kind":"REWORK","_number":2,"created":"2021-03-15 20:11:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/778965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/778965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/778965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/778965/2"}}},"commit":{"parents":[{"commit":"c9e96c9c20280184090c82b46571ee6c69f0cadc","subject":"Merge \"Adding a tool to catch projects that missed branching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9e96c9c20280184090c82b46571ee6c69f0cadc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-05 15:20:25.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-15 20:11:06.000000000","tz":-420},"subject":"Release final python-manilaclient for Wallaby","message":"Release final python-manilaclient for Wallaby\n\nThis proposed a release for python-manilaclient to\nget any final merged changes\nout prior to the lib freeze for Wallaby going into\neffect.\n\n$ git log --oneline --no-merges 2.4.0..8d71f1e\n1bb3287 Correctly pass mTLS certificate/key to HTTPClient\nb5b8bdb Add \"--wait\" option for force-deleting a share/snapshot/share-instance\n9bc7cec Change public share tests\nc225f0a [OSC] Implement user messages commands\nca65447 tox miniversion update to 3.1.1\na34aca6 Add profiling support to manilaclient\nfc5b6c5 Forbid users to create shares with the name \"None\"\n17e5d63 [OSC] Implement Share Revert Command\nc7bcec2 Implement openstack share properties show command\nf1abf30 [OSC] Add the option to explicitly set share status\n08fb635 Tighten asserts in some wait tests\n153aa61 Add a waiter to share extend/shrink\n3c65e6f [OSC] Implement Share Export Location Commands\nd832baa [OSC] Implement the second part of share snapshots commands\n6b1a2f3 remove unicode from code\nbf3e7cb [OSC] Implement Share Adopt \u0026 Abandon Commands\nb77a8c1 Replace deprecated UPPER_CONSTRAINTS_FILE variable\n07898fa Fixes create share from snapshot in OSC\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I55b40d00a58128a652560aba86d0be5712712ad9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fc1604520aad4eb43f5e547963ea16bee620a2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fc1604520aad4eb43f5e547963ea16bee620a2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
