)]}'
{"id":"openstack%2Freleases~779216","triplet_id":"openstack%2Freleases~master~I5ad591ecf2f1269724057e273b244c9d24d72c94","project":"openstack/releases","branch":"master","topic":"move-to-independent","hashtags":[],"change_id":"I5ad591ecf2f1269724057e273b244c9d24d72c94","subject":"Propose to move blazar-nova to independent","status":"ABANDONED","created":"2021-03-08 11:10:15.000000000","updated":"2021-03-08 13:16:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e01cb261698b1b5420c0fcb6c80629150c815b57","_number":779216,"virtual_id_number":779216,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 11:23:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 11:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 11:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 12:44:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"be25b2868a2e198fe1024c988fdcd8edab3b5e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 11:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb3d0fee450e8bafac76ec6f22a2bf9c1dee9c1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 11:11:26.000000000","message":"Topic move-to-independent-blazar-nova removed","accounts_in_message":[],"_revision_number":1},{"id":"148dfb288475c4de440a15d45dc9bbf1975a8f9f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 11:11:57.000000000","message":"Topic set to move-to-independent","accounts_in_message":[],"_revision_number":1},{"id":"36d6013bcd0c411066548234a116cb5ea7be9320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:23:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2760ea905cb94f4eaa778cbf253f580d : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a1aa346acdb4ae892663cfd54dbb9ad : SUCCESS in 5m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/de5af08bc8d64347ba8731e8e7e4fb96 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3c28532e1cd07241b25847fca017418be4c382a8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-08 12:44:11.000000000","message":"Patch Set 1:\n\nI\u0027d rather keep the library aligned with the release model of its parent projects.","accounts_in_message":[],"_revision_number":1},{"id":"e01cb261698b1b5420c0fcb6c80629150c815b57","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 13:16:58.000000000","message":"Abandoned\n\nC.f Thierry\u0027s comment","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe031e8366ef80a59b30023af18b822ad3ab25c2","revisions":{"fe031e8366ef80a59b30023af18b822ad3ab25c2":{"kind":"REWORK","_number":1,"created":"2021-03-08 11:10:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/16/779216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/16/779216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/779216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/779216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/779216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/16/779216/1"}}},"commit":{"parents":[{"commit":"8dda5035850c55511c2f17cfbbabe0f9aca57a93","subject":"Merge \"Release ipa-builder 2.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8dda5035850c55511c2f17cfbbabe0f9aca57a93"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 11:01:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 11:09:35.000000000","tz":60},"subject":"Propose to move blazar-nova to independent","message":"Propose to move blazar-nova to independent\n\nProject did not have any change merged over the cycle, maybe it is time to\ntransition it independent.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\ntransition and we will process right away. If the team doesn\u0027t agree, leave a\n-1 with some detail on the status, then we will cut the Wallaby stable branch\nfor this project.\n\nChange-Id: I5ad591ecf2f1269724057e273b244c9d24d72c94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe031e8366ef80a59b30023af18b822ad3ab25c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe031e8366ef80a59b30023af18b822ad3ab25c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
