)]}'
{"id":"openstack%2Freleases~779234","triplet_id":"openstack%2Freleases~master~If26d523ff98bad06d57ffb49d6f554171f9972ba","project":"openstack/releases","branch":"master","topic":"move-to-independent","hashtags":[],"change_id":"If26d523ff98bad06d57ffb49d6f554171f9972ba","subject":"Propose to move python-aodhclient to independent","status":"ABANDONED","created":"2021-03-08 12:02:47.000000000","updated":"2021-03-08 13:16:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03e142bf4f032e17cdea6bb07403692da6b69afe","_number":779234,"virtual_id_number":779234,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 12:17:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 12:04:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 12:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 12:37:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"152f890d3d7ff9041705423929c15a3f7d118641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 12:02:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b79f94efc4f84d7569a48bf639b6564bbb7e95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 12:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d261a432a30480090531ec193d558d8 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76cf196a5a6d4aba9fd9d3ba63fc1c52 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/617dcc45cdc241dea6c7936f2be6bc0d : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3a3f720f1a216c382b773f50298dddce98bb9c4f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-08 12:37:27.000000000","message":"Patch Set 1:\n\nNot a big fan of moving python-*client to independent, as I suspect it will bring more pain down the line than gains. It sounds appropriate to release client libraries with the same model as their parent project.","accounts_in_message":[],"_revision_number":1},{"id":"03e142bf4f032e17cdea6bb07403692da6b69afe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 13:16:00.000000000","message":"Abandoned\n\nC.f Thierry\u0027s comment","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3ef7145c158ed8271198734a10ebf7cadc9928f","revisions":{"a3ef7145c158ed8271198734a10ebf7cadc9928f":{"kind":"REWORK","_number":1,"created":"2021-03-08 12:02:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/34/779234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/779234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/779234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/779234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/779234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/779234/1"}}},"commit":{"parents":[{"commit":"8dda5035850c55511c2f17cfbbabe0f9aca57a93","subject":"Merge \"Release ipa-builder 2.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8dda5035850c55511c2f17cfbbabe0f9aca57a93"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 12:02:12.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 12:02:12.000000000","tz":60},"subject":"Propose to move python-aodhclient to independent","message":"Propose to move python-aodhclient to independent\n\nProject did not have any change merged over the cycle, maybe it is time\nto transition it independent.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\ntransition and we will process right away. If the team doesn\u0027t agree,\nleave a -1 with some detail on the status, then we will cut the Wallaby\nstable branch for this project.\n\nChange-Id: If26d523ff98bad06d57ffb49d6f554171f9972ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3ef7145c158ed8271198734a10ebf7cadc9928f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3ef7145c158ed8271198734a10ebf7cadc9928f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
