)]}'
{"id":"openstack%2Freleases~779245","triplet_id":"openstack%2Freleases~master~I01778d2848bdb2fba9dd10f513e19234e6e189f3","project":"openstack/releases","branch":"master","topic":"move-to-independent","hashtags":[],"change_id":"I01778d2848bdb2fba9dd10f513e19234e6e189f3","subject":"Propose to move python-solumclient to independent","status":"ABANDONED","created":"2021-03-08 12:18:43.000000000","updated":"2021-03-08 13:09:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b79ea35e2324d7ffa65020e2d2d7f6a6dcad577d","_number":779245,"virtual_id_number":779245,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 12:32:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 12:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 12:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 12:38:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"a0ab2faf8acf5e56ab8ba849cac32f862a303c7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 12:18:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abcc223c3dbd16e17ccd1a9cbbcb1477fbb29941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 12:32:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea711d83ff8b4977b926c42f093e66ae : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b86fb54074664cb7b310e0dccc787622 : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/38b5ff882e0449bf987a91a68d1dd6b2 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c51064f67c00999810b3837cd6e73ea0eaf44ece","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-08 12:38:16.000000000","message":"Patch Set 1:\n\nNot a big fan of moving python-*client to independent, as I suspect it will bring more pain down the line than gains. It sounds appropriate to release client libraries with the same model as their parent project.","accounts_in_message":[],"_revision_number":1},{"id":"b79ea35e2324d7ffa65020e2d2d7f6a6dcad577d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-08 13:09:22.000000000","message":"Abandoned\n\nC.f Thierry\u0027s comment","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1f54728dfa5387cd33a6b73123a66d9886c67f5","revisions":{"e1f54728dfa5387cd33a6b73123a66d9886c67f5":{"kind":"REWORK","_number":1,"created":"2021-03-08 12:18:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/45/779245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/779245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/779245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/779245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/779245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/779245/1"}}},"commit":{"parents":[{"commit":"8dda5035850c55511c2f17cfbbabe0f9aca57a93","subject":"Merge \"Release ipa-builder 2.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8dda5035850c55511c2f17cfbbabe0f9aca57a93"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 12:18:23.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-08 12:18:23.000000000","tz":60},"subject":"Propose to move python-solumclient to independent","message":"Propose to move python-solumclient to independent\n\nProject did not have any change merged over the cycle, maybe it is time\nto transition it independent.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\ntransition and we will process right away. If the team doesn\u0027t agree,\nleave a -1 with some detail on the status, then we will cut the Wallaby\nstable branch for this project.\n\nChange-Id: I01778d2848bdb2fba9dd10f513e19234e6e189f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1f54728dfa5387cd33a6b73123a66d9886c67f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1f54728dfa5387cd33a6b73123a66d9886c67f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
