)]}'
{"id":"openstack%2Freleases~780221","triplet_id":"openstack%2Freleases~master~Ib7ea93fce7e618a2dbba8dd9f4153897bdfee22a","project":"openstack/releases","branch":"master","topic":"standardize-gerrit-topics","hashtags":[],"change_id":"Ib7ea93fce7e618a2dbba8dd9f4153897bdfee22a","subject":"Standardize topics to use during series","status":"MERGED","created":"2021-03-12 10:58:10.000000000","updated":"2021-03-25 08:57:41.000000000","submitted":"2021-03-25 08:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"780221-standardize-gerrit-topics","meta_rev_id":"69550e6a815ff06e3567294a636cfe1b2809494c","_number":780221,"virtual_id_number":780221,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 08:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-25 08:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-24 12:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-25 08:47:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 11:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 10:13:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-03-23 17:29:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4fa23f2e26b40e46a467e4c140879c3fa58e856e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-12 10:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52f303530acc6b8a407f205fa1aa000fdc55b2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 11:12:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac15b39dc46a453e8d5336bd52d64851 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6957668c9b9186dc2c0f57e7e28b35aa9ef3456f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-15 10:13:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfcd361180a61bdfd2c0512a7bf203baa965a031","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-23 17:29:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)\n\nSome wording issue inline.","accounts_in_message":[],"_revision_number":1},{"id":"997ebacda1b151b91f8325de708b208fc25627f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:39:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8e2fb1e15c0f32109268c15985d920d6c938642","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:39:58.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (8 comments)\n\u003e \n\u003e Some wording issue inline.\n\nIndeed, thanks Elod.","accounts_in_message":[],"_revision_number":1},{"id":"d8b6f330fe200b4e7ac90387823a5fdd62ff0259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 09:52:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d19e7b5d540149e984b18407190c8ab9 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6279c1a2a3a786a4af5b2cff759e3fd21ac623c8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-24 12:12:50.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, too, Hervé!","accounts_in_message":[],"_revision_number":2},{"id":"c2aff1c02d471688aa0feb3c249cd423a8905b22","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-25 08:47:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9d55263001be3c2a330a399dcf5cc5a2e656efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:48:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8da11f20d8b3c39ee222bf497511567a859b24ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:57:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4585ef5725be4a09b9b60b219ab98125 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0409e2eb48d5471e91f485e8a8a17321 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"69550e6a815ff06e3567294a636cfe1b2809494c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:57:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1bdf3acb51c38f3006a7651f8e361e9ec49c85c","revisions":{"87ededa3a1e1ac074ee99c9696dfed36eef17e4e":{"kind":"REWORK","_number":1,"created":"2021-03-12 10:58:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/21/780221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/780221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/780221/1"}}},"commit":{"parents":[{"commit":"c4b3492f630880f515223cd24e2bf33f8a8c25c2","subject":"Merge \"Move futurist to independent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4b3492f630880f515223cd24e2bf33f8a8c25c2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-12 10:52:02.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-12 10:52:02.000000000","tz":60},"subject":"Standardize topics to use during series","message":"Standardize topics to use during series\n\nThese changes aim to standardize the gerrit topic to use with on our patches.\n\nThis:\n- will allow us to better retrieve the history from previous series;\n- could allow us to generate some statistic and monitoring from series\n  to series to allow us to get a better big picture of things are.\n\nI already defined this point the potential process/tools changes for Xena.\n\nhttps://etherpad.opendev.org/p/xena-ptg-os-relmgt\n\nMaintenance mode is becoming more and more prominent, a little standardization\ncan\u0027t hurt.\n\nChange-Id: Ib7ea93fce7e618a2dbba8dd9f4153897bdfee22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87ededa3a1e1ac074ee99c9696dfed36eef17e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87ededa3a1e1ac074ee99c9696dfed36eef17e4e"}]},"branch":"refs/heads/master"},"a1bdf3acb51c38f3006a7651f8e361e9ec49c85c":{"kind":"REWORK","_number":2,"created":"2021-03-24 09:39:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/21/780221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/780221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/780221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/780221/2"}}},"commit":{"parents":[{"commit":"c4b3492f630880f515223cd24e2bf33f8a8c25c2","subject":"Merge \"Move futurist to independent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4b3492f630880f515223cd24e2bf33f8a8c25c2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-12 10:52:02.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-24 09:39:34.000000000","tz":60},"subject":"Standardize topics to use during series","message":"Standardize topics to use during series\n\nThese changes aim to standardize the gerrit topic to use with on our patches.\n\nThis:\n- will allow us to better retrieve the history from previous series;\n- could allow us to generate some statistic and monitoring from series\n  to series to allow us to get a better big picture of things are.\n\nI already defined this point the potential process/tools changes for Xena.\n\nhttps://etherpad.opendev.org/p/xena-ptg-os-relmgt\n\nMaintenance mode is becoming more and more prominent, a little standardization\ncan\u0027t hurt.\n\nChange-Id: Ib7ea93fce7e618a2dbba8dd9f4153897bdfee22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1bdf3acb51c38f3006a7651f8e361e9ec49c85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1bdf3acb51c38f3006a7651f8e361e9ec49c85c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
